)]}'
{"id":"openstack%2Fosprofiler~872152","triplet_id":"openstack%2Fosprofiler~master~I743fdc1203af6735d5772728e1b5c2ca16d41d9e","project":"openstack/osprofiler","branch":"master","topic":"jaeger","attention_set":{},"removed_from_attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2023-02-16 14:14:05.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-02-16 12:02:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-02-16 12:34:39.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"I743fdc1203af6735d5772728e1b5c2ca16d41d9e","subject":"jaeger: introduce service name prefix","status":"MERGED","created":"2023-01-30 10:54:23.000000000","updated":"2023-02-16 14:15:14.000000000","submitted":"2023-02-16 14:14:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872152-jaeger","meta_rev_id":"7e9a6d02dda47422ca8cba96c98ad89f692be272","_number":872152,"virtual_id_number":872152,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-16 14:14:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-02-16 12:02:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-02-16 12:34:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-02-16 12:35:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"date":"2023-02-16 12:35:42.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-30 12:33:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-09 13:06:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-02-13 10:51:23.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"17e1fa8431ae15cb282c57d1d9b5dd4830324e08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-30 10:54:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1650b79de5b98ddf8bf356555187be2799070880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 12:33:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5206aff993cd4ab6aa77dffb1980120d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8880c0f6772344e18c8e8702bb58e3ab : SUCCESS in 1h 25m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db4f63903c6e4ab59a3ed72200c70c72 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7ff4812a4b34234a015e2febea54ee7 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9de16ec55284612beafd3c5bcfaa736 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b4d3e24252d42b093fd4322af4c5fb1 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc41ab70f3c4504b420ae956efed72f : FAILURE in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c019f34cce94c3cab7286b229c51087 : SUCCESS in 3m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32f7de3051ed430da4c44cc6899d5d8f : FAILURE in 3m 04s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/604ee71d00a34644a410fc322da61207 : SUCCESS in 28m 19s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/04e83df53708414f919f4b1473b56aeb : SUCCESS in 27m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a08b32cdc7d3e131a120efca15006b66c429ce2","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-30 12:55:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e82eb63ad798d4e5b6cd9c7ee277c27e137437c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 15:07:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14662e4a3d4b4006af88db2325b331b2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c155195415054e55bfed5ad241c30d30 : TIMED_OUT in 2h 05m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a20cc9ab4672467f9c4a58c586783b09 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a387e06936946dcb4c1b973bcd3364f : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1ed48b83ffd47118acfb2bc454ed2db : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e3904a8c2524d018cf2a443a7b1e6a7 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ce7c9342dca48c4ab109ae758782589 : FAILURE in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2736b4cf4c1c4360873d59c703e49ea4 : SUCCESS in 2m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b26cf927011b44feb23269133436ad57 : FAILURE in 2m 59s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/aeeb20efb84c4443b44948185224a947 : SUCCESS in 43m 28s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/253fbc3590184237b1f742e5704de872 : SUCCESS in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69e494aa582da114802a7798176536175a583514","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-01-30 15:21:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbf7e548e8cdf45656010712f2e7d89e0b83d637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-30 17:01:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dddc3c8c68fc447086fc2028592ef3d1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e40eea031ecc42798e14efbd035ed572 : SUCCESS in 1h 38m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dabfb1005ea40bd9d0ea6d34b9e59dd : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2548b87eb944129b9430f9b8ec34941 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/793900dcff60495985644608dd6f77c2 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71a80af36eec44ec8e72b94df4ea7b80 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d133371484d44a5a020a0b9f9229903 : FAILURE in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2791b1b26754f0d9b46a0183e7d41d9 : SUCCESS in 3m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6bd3f0a6845f4d96880d1d9a47829100 : FAILURE in 2m 32s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/50931668bae5401691edf38985ee5b39 : SUCCESS in 32m 59s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/942daa10b7e84c61a7dd018d280a9369 : SUCCESS in 29m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7053d35d22b14065de5d19e7fbbdeb0eed85fb41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-08 10:52:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"25048c7ba203103b2bb56b0a0d9f38d5acb1d739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 12:45:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f601646ac31d4b8d87942413485f3b91\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/567d4529cd7846028282583a88279e96 : SUCCESS in 1h 46m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5257b13d1f2d4753bfd1466bdd450113 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb25e7d6229044ceadcfcaccf30b1543 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0802f53ad60441098cc50c01229624a : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25145a3502014da19a78b80d1d676572 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd088220a60d49d28a4bfb5a26de9dbf : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e9f5435c04941839b1950f6fdab3c21 : SUCCESS in 3m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f18189857ce34ebeabff1e0f9fbfa645 : SUCCESS in 4m 19s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/fd477590b9ce4a72b5aac3fcc910628d : SUCCESS in 32m 37s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/f759efcdf0704df38442ca5752ad72bc : SUCCESS in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cbf3784bac6c17f94d50b073ede987b4fe8198d4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-09 13:06:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7493e6d2624d70deb7c20cf97cc6f8362a13777c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-13 11:57:14.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3dd10493fa4277a6abe61b6f8cc64b2d5c792d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-13 12:09:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"84f47e7d228bb710aeb236d2ee2bd412cbe37e5b","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-13 12:10:11.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9bbb232b98e1d605349c3300f0f681caa184c035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 13:58:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03bdf241f754431392f61100dbb77977\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c459c59249b34a718c09e09454268e08 : SUCCESS in 1h 40m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffc035939d494d3f83339dede6b4e20e : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a324e04d4684997b598a23dbc9198a0 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88fdf58d793d4dde81d2e619e80d3a65 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42dc837c7d6540f7964a82762b1acc08 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1dc021b3a8441b4a315e31b8bcb0149 : SUCCESS in 3m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/528fe4c01afd4d0fb4b51a881ac2d643 : SUCCESS in 2m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e44e63d6208f4c37b263a6f23c0860f4 : SUCCESS in 3m 27s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/08d3e4efdfa34cb6bbc131feccbec076 : SUCCESS in 39m 59s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/fec03089cf824f25b30fd49041caecf5 : SUCCESS in 43m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8e3edc010b549caada8c4705e9ec150ae50e913a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-13 14:04:05.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f0054a2f7f698868da03b09a67d30827f96f2b9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-13 14:12:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f1f2783564166e31a0dfeee26979480b34f53e75","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-13 14:17:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"09a4f7c44ce99763088f4a5eabda2411595ed7bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 15:47:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5944ef7b0f849d6bac760b5bdb652ce\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d11c8122c2f42af830f451323a14412 : SUCCESS in 1h 31m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f64eb9a08df4133af44dfa4159d0ff9 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3afdc5adaf474e9b8f192497da988642 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e07ef1f6f6a74da48a413c39ee1f082f : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/939d8954f6ae4a9fa31f485f398dcce6 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2070be1184774878bae7cb18368c7dc4 : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/509fc02adf9e41ba8284017dd67d3847 : SUCCESS in 3m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2e3d091251804d3fa7141da0f0f1597d : SUCCESS in 3m 32s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/abd7a67844414ea7b04ab82db705b1c4 : SUCCESS in 44m 45s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/38a045f631e14a94a5e13461dee76a0c : SUCCESS in 56m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3fd040cb85b766b593a945ebeccb03df967020b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-13 20:01:01.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"213afe5e7db24c123f61363505a3ab0e04a854a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-13 21:30:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b91163f0db640a69bac82fbda947e8b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88f4cb7dd3e246f2a51b8731c0c3e16e : FAILURE in 1h 27m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f86cdbb9ada4e198c0aeb43c99adcaf : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8787ef96dbab41a0850a78cca6c7cabd : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81f7f693365c4e909e3893801ba439c8 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d663efda8c384ee79c59e408dbee9a5d : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76345ea47df94b1db94e1e8949b104aa : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18ca88a55df145e9a620e2bcc519666b : SUCCESS in 3m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/09d9439c51434c1cb3e9678156ad76ee : SUCCESS in 3m 30s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/4538fb89c3424888b7208b347419206b : SUCCESS in 27m 09s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/2c38fd41835844ccaf015ab6a0342878 : SUCCESS in 44m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"21d67f0c195356c922fcc18c302486768c340594","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-14 07:48:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"756e7dc10eee89560c277ee11be2c6768cb8da5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-14 08:37:55.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"11b0e74ca3e8df9dc6f51258f28280a621aeb1ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-14 10:36:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a68a2f18d63e40b296ccf4309d209eea\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/305f3882eac1468ca6fbdf3ece32450c : SUCCESS in 1h 42m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8baee13a6b924c6cbac6a94a7fa3d44d : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edd9da14705d4b2dab8946c124bd80c0 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c3c858a116b4eac89a898cd895d4cfa : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01087f0d549e4373b1d6e41b3f6e4053 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/684a230a680b4536a0da78e6364eaa74 : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae4f6afc54954f8ca2095cdf2db4126a : SUCCESS in 3m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/650583a896824919ad4b2872a8d8b4f3 : SUCCESS in 3m 18s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/82bb4dbdd716413f86415f648d1cb29d : SUCCESS in 34m 51s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/09c0d5a5f4eb417b83325ebaa9dea7da : SUCCESS in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"78a6141e2bf403d5ec8ca885a8ac38df67d41473","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-15 02:02:39.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"003c608fbd2505f55638357274570ebc33d291b2","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-15 07:58:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7a209489bf8070f98381d5a87b57784d3ff41467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-15 08:00:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e519fe868b26b4b480bc626e32701170765b4631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 09:47:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64be1d272653469c97de6fb5277642d9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1dd064e270446beaf3bb2aa1e396ad2 : SUCCESS in 1h 40m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2646ed19091e44689a8460bc14531fad : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9487b418dd0a4cbb8726d03d4649c665 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2206cdcdce654cb7aac1c2a2c0555b82 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59616c7de76d444fbe106d39e2803aa3 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/005fd8d6f2ff4ffeb109b38b99d372cf : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34244d00db5c48bc9ad9599dc4c3007f : SUCCESS in 3m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9b0225c3f2bc4f24abecdce5f5b66a41 : SUCCESS in 3m 37s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/f3cb1cc0525c4ab8bd8a27aab1e5f8a2 : SUCCESS in 32m 40s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/d35f23bee90d4cb996e3e4dbc6711933 : SUCCESS in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"999f415bda20cde31b9d4b09a2d3b9d93b6a2c63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-15 16:52:21.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"fdcaf3a5de3409a7ebf8446986491db94cfc59ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-15 18:32:04.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e2b261f9f324495a4f3b2b0e654f08a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/84612a1658284221a75c757e554d868b : SUCCESS in 1h 30m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e907e83ac474d4a8b8878dff3725e93 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6882db1391cb4c34918d7af1231fb92f : FAILURE in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e1805c686fa4a3a931b3868a9df207c : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f213bc17682c4990b0ae05799ab874a0 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee018026362845adbfbc8e684335a347 : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c11022bfe1464bfca7232f8a2fc8b62e : SUCCESS in 3m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f3dd313c1c6e41b48b8ef514d27034a6 : SUCCESS in 3m 40s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/2ed2779b04bc461e9909eae69aaa6051 : SUCCESS in 39m 37s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/054c360e93f34b588e8803f5a1838b33 : SUCCESS in 39m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1e627ffea4bd91531d3931b77b57a641e8d5eaac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-16 08:09:01.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"5d78aed4126186f019177580cd5160a89736ea49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 09:33:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/268238045b2541c09369f6897a8715e6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc794bf1b3f1426194c1c6835cec442e : SUCCESS in 1h 23m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a942c23815834a2e9170344bb6ff8c0e : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0766ccfa34094f4d9d588cc165f02da0 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0e50347e1284cb98beedbd13ffb2115 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6804dbf47af8464ab479cc3a811f784b : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de22a7299e37472bb7e224ecfc2fdae8 : SUCCESS in 2m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca4c86f454164995ada9d861995a6680 : SUCCESS in 3m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f21a6c3c72fe41649ff276b866c2b032 : SUCCESS in 3m 05s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/19fb210cd7ab4965ab30cbeb4967c4e4 : SUCCESS in 34m 12s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/913e2e04fd6f49f0b41d054838d03c6e : SUCCESS in 40m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"0a8ab115eb1311594987ac33fab76905898e3347","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-16 10:57:39.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"768fd86a4751c40eac486b6ee1438df063d84324","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-02-16 11:08:49.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"471d0ba2574a0e79c94904bea10701223ccee9af","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-02-16 12:02:05.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"66f6d0f10915c79527826736eb3e2140cc0a0749","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-16 12:34:39.000000000","message":"Patch Set 9: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4abe92a934a596c9be5cb7a90af2724e49d844fe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-02-16 12:35:42.000000000","message":"Patch Set 9: -Backport-Candidate Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"51e372f219ce92c435b78036bed37dbbb32e7fbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 12:36:19.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"3a8bdfe8005f56990ef13ab2056364a98e0c81b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:14:05.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ba42f4288fe42ebbce0b066eecd676b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e02c309914d242c1a9b434ee81284f6a : SUCCESS in 1h 32m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b2f084e5085484eabf3f07c57b01641 : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e8c6ef7b1a44c44b1b79aa807bc7c97 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca7346daafb94796a4ddd6bc85312a3c : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb27972656a04e069105b241cfc97da6 : SUCCESS in 3m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab5a1a7e5aa94296a1def6197ada6486 : SUCCESS in 3m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9b47ac64d0704a6d88ac0d10536036e1 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":9},{"id":"d4b93c30ce3d1741f35870409d11f917b7dba608","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:14:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"7e9a6d02dda47422ca8cba96c98ad89f692be272","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:15:14.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9135c8ffdbb4818b336b7917991f3d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649820ba79f049b29986a5f3ebdcd82d : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82a6293d21e94f4ca1ee605f81dec965 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b7468756bcfdf3fe74e7cacb0562a7f969472250","revisions":{"67cdeeaef19dd9d0f88c8a63e04315b3fc614a0f":{"kind":"REWORK","_number":1,"created":"2023-01-30 10:54:23.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/52/872152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/52/872152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/1"}}},"commit":{"parents":[{"commit":"0b1bfadebdeacda93c117fa0ab513bc228039372","subject":"remove unicode prefix from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0b1bfadebdeacda93c117fa0ab513bc228039372"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-30 10:53:18.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-30 10:54:22.000000000","tz":60},"subject":"jaeger: introduce service name prefix","message":"jaeger: introduce service name prefix\n\nCloses-bug: #2003092\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I743fdc1203af6735d5772728e1b5c2ca16d41d9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/67cdeeaef19dd9d0f88c8a63e04315b3fc614a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/67cdeeaef19dd9d0f88c8a63e04315b3fc614a0f"}]},"branch":"refs/heads/master"},"413f6e013ee36ec1a0d079581ffa6417b922ff17":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-02-08 10:52:41.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/52/872152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/52/872152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/2"}}},"commit":{"parents":[{"commit":"be6aa856a5385d7a9dbf04683ed6a263f12de48e","subject":"Merge \"tox: Add functional-py38, functional-py39 envs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/be6aa856a5385d7a9dbf04683ed6a263f12de48e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-30 10:53:18.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-02-08 10:52:41.000000000","tz":0},"subject":"jaeger: introduce service name prefix","message":"jaeger: introduce service name prefix\n\nCloses-bug: #2003092\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I743fdc1203af6735d5772728e1b5c2ca16d41d9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/413f6e013ee36ec1a0d079581ffa6417b922ff17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/413f6e013ee36ec1a0d079581ffa6417b922ff17"}]},"branch":"refs/heads/master","description":"Rebase"},"0cea654a638462e95190b2d5f5053d4edd215a31":{"kind":"REWORK","_number":3,"created":"2023-02-13 12:09:43.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/52/872152/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/52/872152/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/3"}}},"commit":{"parents":[{"commit":"be6aa856a5385d7a9dbf04683ed6a263f12de48e","subject":"Merge \"tox: Add functional-py38, functional-py39 envs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/be6aa856a5385d7a9dbf04683ed6a263f12de48e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-30 10:53:18.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-02-13 12:06:54.000000000","tz":60},"subject":"jaeger: introduce service name prefix","message":"jaeger: introduce service name prefix\n\nCloses-bug: #2003092\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I743fdc1203af6735d5772728e1b5c2ca16d41d9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0cea654a638462e95190b2d5f5053d4edd215a31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0cea654a638462e95190b2d5f5053d4edd215a31"}]},"branch":"refs/heads/master"},"1c2c2b9f6a087b7a62b8f2da3231e8bfa6859bab":{"kind":"REWORK","_number":4,"created":"2023-02-13 14:12:42.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/52/872152/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/52/872152/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/4"}}},"commit":{"parents":[{"commit":"7d2c871b7167f193850161f0188d74daf02ad8e0","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7d2c871b7167f193850161f0188d74daf02ad8e0"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-30 10:53:18.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-02-13 14:12:36.000000000","tz":60},"subject":"jaeger: introduce service name prefix","message":"jaeger: introduce service name prefix\n\nCloses-bug: #2003092\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I743fdc1203af6735d5772728e1b5c2ca16d41d9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/1c2c2b9f6a087b7a62b8f2da3231e8bfa6859bab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/1c2c2b9f6a087b7a62b8f2da3231e8bfa6859bab"}]},"branch":"refs/heads/master"},"a12939bee1b66926fc464dff5fb4e24547e42e45":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-02-13 20:01:01.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/52/872152/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/52/872152/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/5"}}},"commit":{"parents":[{"commit":"512ff03eeb04fa5ef64c3ee5b799aeac1b7d89e9","subject":"jaeger: fix driver initialization for tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/512ff03eeb04fa5ef64c3ee5b799aeac1b7d89e9"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-30 10:53:18.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-02-13 20:00:08.000000000","tz":60},"subject":"jaeger: introduce service name prefix","message":"jaeger: introduce service name prefix\n\nCloses-bug: #2003092\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I743fdc1203af6735d5772728e1b5c2ca16d41d9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a12939bee1b66926fc464dff5fb4e24547e42e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a12939bee1b66926fc464dff5fb4e24547e42e45"}]},"branch":"refs/heads/master"},"7fead31c8fe17102bf8fcb77fca9167a0e220993":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-02-14 08:37:55.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/52/872152/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/52/872152/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/6"}}},"commit":{"parents":[{"commit":"a7d857d4753943dc4f46900dd4f8c3c814a25067","subject":"jaeger: fix driver initialization for tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a7d857d4753943dc4f46900dd4f8c3c814a25067"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-30 10:53:18.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-02-14 08:37:35.000000000","tz":60},"subject":"jaeger: introduce service name prefix","message":"jaeger: introduce service name prefix\n\nCloses-bug: #2003092\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I743fdc1203af6735d5772728e1b5c2ca16d41d9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7fead31c8fe17102bf8fcb77fca9167a0e220993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/7fead31c8fe17102bf8fcb77fca9167a0e220993"}]},"branch":"refs/heads/master"},"fcb211196c25ba3a6b133b045e896b2aeca020f6":{"kind":"REWORK","_number":7,"created":"2023-02-15 08:00:18.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/52/872152/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/52/872152/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/7"}}},"commit":{"parents":[{"commit":"a7d857d4753943dc4f46900dd4f8c3c814a25067","subject":"jaeger: fix driver initialization for tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a7d857d4753943dc4f46900dd4f8c3c814a25067"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-30 10:53:18.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-02-15 07:59:59.000000000","tz":60},"subject":"jaeger: introduce service name prefix","message":"jaeger: introduce service name prefix\n\nCloses-bug: #2003092\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I743fdc1203af6735d5772728e1b5c2ca16d41d9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/fcb211196c25ba3a6b133b045e896b2aeca020f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/fcb211196c25ba3a6b133b045e896b2aeca020f6"}]},"branch":"refs/heads/master"},"c7b9a5738efd406db7b58ee6cce649121cbe981b":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-02-15 16:52:21.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/52/872152/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/52/872152/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/8"}}},"commit":{"parents":[{"commit":"d693b2a950a69f54afc79590849606d468832899","subject":"jaeger: fix driver initialization for tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/d693b2a950a69f54afc79590849606d468832899"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-30 10:53:18.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-02-15 16:51:30.000000000","tz":60},"subject":"jaeger: introduce service name prefix","message":"jaeger: introduce service name prefix\n\nCloses-bug: #2003092\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I743fdc1203af6735d5772728e1b5c2ca16d41d9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/c7b9a5738efd406db7b58ee6cce649121cbe981b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/c7b9a5738efd406db7b58ee6cce649121cbe981b"}]},"branch":"refs/heads/master"},"b7468756bcfdf3fe74e7cacb0562a7f969472250":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-02-16 08:09:01.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/52/872152/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/52/872152/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/52/872152/9"}}},"commit":{"parents":[{"commit":"9f50374dde1362cf2c1d63d2e98d5b89b983a391","subject":"jaeger: fix driver initialization for tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/9f50374dde1362cf2c1d63d2e98d5b89b983a391"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-01-30 10:53:18.000000000","tz":60},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-02-16 08:08:50.000000000","tz":60},"subject":"jaeger: introduce service name prefix","message":"jaeger: introduce service name prefix\n\nCloses-bug: #2003092\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I743fdc1203af6735d5772728e1b5c2ca16d41d9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/b7468756bcfdf3fe74e7cacb0562a7f969472250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/b7468756bcfdf3fe74e7cacb0562a7f969472250"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
