)]}'
{"id":"openstack%2Fosprofiler~879318","triplet_id":"openstack%2Fosprofiler~master~I00e9f5661a3bd54acc846e8c326896d21e70be36","project":"openstack/osprofiler","branch":"master","topic":"opentelemetry","attention_set":{},"removed_from_attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2023-06-06 09:59:07.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-06-06 08:19:34.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2023-06-06 09:59:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I00e9f5661a3bd54acc846e8c326896d21e70be36","subject":"profiler: add python requests profile","status":"MERGED","created":"2023-04-03 09:28:07.000000000","updated":"2023-06-06 10:00:09.000000000","submitted":"2023-06-06 09:59:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"879318-opentelemetry","meta_rev_id":"3ae10e6f7fcbb1ecb2f722e16a33e227fdc4fd35","_number":879318,"virtual_id_number":879318,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-06 09:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-06-06 08:19:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-06-01 13:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-06-06 08:19:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 10:53:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-11 12:21:21.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-05-23 02:58:52.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-05-23 08:55:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2023-06-01 13:35:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"673c71f8c74e43aee7f015fd2b9e6a2f5cb3a3a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-03 09:28:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ca97a33dbad908d050fb437cad5f33576e06ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 10:53:15.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c415496302e4370a5f58be54fd257a2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/feb2091d20b9456abd2940027aa10c49 : SUCCESS in 1h 19m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/711319b1d7a8427d87df532229ad9595 : FAILURE in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26e7ab6b02824154b6a5cfc878292f64 : FAILURE in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b94529ef0c02471b90f1122d4b65605a : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03601b23ac8d477aba09622fa1a8441f : FAILURE in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e90f1baff904164bf48b63b35e3d8e7 : SUCCESS in 3m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5a10f9c3d214430bc8cea046348405a : SUCCESS in 2m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7d9169de56b74eeda4de149cccc12288 : SUCCESS in 3m 30s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/660aa4e9d50b47e087c1b8ac0b9871a8 : SUCCESS in 39m 43s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/355b6b909f234cfaab74f8f401fd1f8a : SUCCESS in 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fbc27ca1868839df1edebc1c30ed9173f341113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-04 08:21:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03624558e7951b86e7ca3dd6f01e9985cd67749e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 09:40:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdbfbef7651b4cb68dccb5fb5a17f60a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c85d72944064558ae039c93c21ba86e : SUCCESS in 1h 16m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eeb1e830beb643b9850332667e8dcd4b : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b95f80960c94c3dbe3d3ca065b05d7b : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2055ea0bf00c44a582fd180885221231 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92bcc78e293b467da6f7d1a127a8ffcd : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9a66a79df044179f51b0623559ca6d : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acdb10566a144ebb8ddf8e9f7fb23a65 : SUCCESS in 2m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/66419c1b76be4c4b85424a08ff575a20 : SUCCESS in 3m 32s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/feebc64a91e14e1ba384e973229cfe58 : SUCCESS in 38m 47s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/9ffcf2ee4ea64944ae948a1eb0670ddc : SUCCESS in 47m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"45bc20dafabed9842651ba1b6169ea742f7e382a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-04 09:43:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6967e61bc1953f024d1c3896db17e33c3398139d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 11:06:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24fcee653b464def966fd0ef1266975e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f091eb9928994a54abd753fbc4b43b66 : SUCCESS in 1h 17m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad3d644079b14ca386338dc0be6d574c : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4b431b244844548ccbf8e3668380be : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fc32d453bc74e4cadd8e1f01189d662 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ce8226037c644b1a491809dd6e671f0 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a91d050c0625440c938d20d97eb7633e : SUCCESS in 3m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f195a83101b54f94bb550b2f5078ba1d : SUCCESS in 2m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3c53b5f79a994e648c2097e28a1ff290 : SUCCESS in 3m 31s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/d5a9a5d48075489b92e88255cf8e565d : SUCCESS in 40m 22s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/7ba3e816e56d4091b6a7e18b7ffc9166 : SUCCESS in 38m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0f84e321b9fcc989a2fbd7ca23804568a3fbfa62","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-11 12:21:21.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"64856e5c860ef37695fead1a0d630532f733ce34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-24 09:52:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1153ec61bea8c2869a4bbe610e886dee1092cbe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 11:31:45.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06fb39c2e19d45228f2385de4d682f94\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d96fb9566fa74b5ca4f4af15a0f29af7 : SUCCESS in 1h 33m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed2659c1ca0f4e51abdfb18b2fa1fd83 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c9db12ed0a444b7a0b30cbb4b829403 : FAILURE in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4059730ea9b044f7968e0633c0e1fde3 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55b9a97a898b474fad0182e57f191364 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95478ada9bb543fdbe6e31f7b6ecf4a2 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/576a9bec1a2947b898d357e11b46cb9b : SUCCESS in 2m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3b20c334985b47c5a36649e624eb5686 : SUCCESS in 4m 36s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/e08982ffc65042c3a88fb96b03736812 : SUCCESS in 42m 08s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/673ed8b84f7c4d9cb4c212022a2251d5 : SUCCESS in 50m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d13fa90175bf04b117d7783342b88c985bcb7add","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-24 13:53:34.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f03468e4bb038e169986b2a2213ffca896fc2530","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-24 15:06:59.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"ff7a3a1fc553c7dfaf3bafd69009fb4084a581fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 16:47:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fcaf0b017594f7a8653431e25509e3e\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/992ac13fe7e647cdb14cfb5a5add4080 : SUCCESS in 1h 32m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc9503a51c9a437b9645c1a7fba95977 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a89f49bfc7b4e9d914096fc6a2b1755 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89badefa3bbc43abab619801365adddf : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82318ddf841f41e7bf1c80c31e549984 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ffbe2214304744842fb229e828ed0e : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd7df1d1f45743229735fd81d7df9997 : SUCCESS in 3m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/92547d9767ea4c9486b002c1bd30d64e : SUCCESS in 4m 30s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/e2a34d596f7e4ba99721085050a2208d : SUCCESS in 25m 52s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/e5fe0557fb46491c8dc0721809b4da3c : SUCCESS in 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"04f81067bc60b1f616d3ee5b12e07c71d6294933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-25 09:08:00.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"9890b5a58349b172babfdf90c685b6cda67bafba","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-25 09:12:42.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3c22580fc10d54677f221f7a92d7a3debfa461b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 09:39:48.000000000","message":"Patch Set 7: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/269fe6e841cb4c1da87bb4aed3119368\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/107fab4c113b4825a4d2b6e2ea78cd36 : FAILURE in 26m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20fd6609cc1a4fb0b965da79a11e0b92 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b445563942f44388143dfdf31729be8 : FAILURE in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e93366bdf0164cf88888a8c4818d7612 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d43c0a6a621a41af90427c581b3765f9 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0690c21ec2e547ebb8143218c6b54987 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01a9ded41fd14a3ea61d5fd57e636c9c : SUCCESS in 3m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bad6fa03f9f7478091b0e9eeeafe277b : SUCCESS in 3m 14s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/0b11749e56dd462e84d7e1ba1b8808a1 : FAILURE in 25m 20s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/c5b8a067b39b4d269a7b79381be2576b : FAILURE in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8ccc68e83f4536838c997e24aba82762be9e69b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-04-25 12:03:50.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"601af1bb8c3f686e2afac31ca7b06e067e68ac57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 12:19:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22c17a16060542e3a19719ae04114a74\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76120798e1b14dc48854ec03238d1538 : FAILURE in 14m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/698845941284448d87c6dd301197b478 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8b68c13026d4da985f3e8bb4efedf6d : FAILURE in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e574d84ba6304321a9e640d82b6bbd3b : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f48e1aef1b88498182e8e83911e40d7a : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1959646b0e4116bafae1f71479c821 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b76e34ded2f464ab12f7143205e7ac8 : SUCCESS in 3m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/93e526fa624f444092c28d38ba469617 : SUCCESS in 3m 41s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/4e36e4b8b14c49de9ef48d3739e7b736 : FAILURE in 14m 11s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/34641c0a732746a89ae21f0b87d49469 : FAILURE in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3f6a24b0d071aac0444b2cc4f334f7417a6fbd2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-03 09:12:11.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"6fc94a3fdf177188a72c3343637a40f8b67c95a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 09:43:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e7e9cf4e7e74b3caf7fc798444af709\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7e03e96cb9548fb8f52c98e3b3dcd19 : FAILURE in 22m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ca82aacc4d74902a276988a08054c89 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc6bdf5eae54e20ada150a17565d1df : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ebbe1f22670461292bbeeff56781f22 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2262769cf094424089fe26ec27578779 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c24c0f76f83407892f94bad2685bed4 : SUCCESS in 3m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf9b31348fa2453a85764d826d1c1de4 : SUCCESS in 2m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6dfd078f6a6f4943877103eafe6ba18e : SUCCESS in 3m 35s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/7272a756a39e4fa8a6907c86eb3c4b51 : FAILURE in 23m 43s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/353efa5d4a094bf1b44f7b1507a41d97 : FAILURE in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"539e77996424e1859070e83aa2bd3abd6a8cd4f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-10 08:40:39.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"7f60cd3a9283859d6a6271aabfba33b1cfc1f6d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 09:07:46.000000000","message":"Patch Set 10: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/897a50990364463889ff9ea9de15ea61\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/205c5428676342ea99a1301086ff9d60 : FAILURE in 22m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2641887fcf74f39a6ec8074407365be : FAILURE in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64ed357bd5ab4c78bea8c51c823095da : FAILURE in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9ac2af9f1dc4a899b0270c042be4101 : FAILURE in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ced0b2f14e4f4485858c9f388f51f2fa : FAILURE in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef69ba70038a498c884cf67393b444c9 : FAILURE in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c9b8158c50f440f9ab01965f4072b53 : SUCCESS in 3m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a32dc2fc3f546e0a4a214d7196a9ad2 : SUCCESS in 3m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a2f8b527af641879c618a96c1da091b : FAILURE in 3m 29s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/043d6554e29a4d39baf9f6a7fb8e75b9 : FAILURE in 13m 21s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/364d162b9f104b0c989ef94690c33d26 : FAILURE in 13m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"801372fb21b97b25e36103ef27405a3634e31758","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-10 09:38:36.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"d962e8b2078f0d68081b972f2825c04a98bad565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 10:59:50.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0ed8dcf065f4738b9d0c40bd9b802a7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d941905f484422ca4189295ffbeab71 : SUCCESS in 1h 19m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d36a70c244484c948a5575ebaea21791 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b65f6df8f14cbdb1691da751ffafea : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc009b92a9094415b291c29c69989f29 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ab1ef39539445f8b6f3c07fff1260e4 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f48da1da2d124e1f90503c20560e9a95 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6835854db60c4d0f8ab65cdc9ab3e79e : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/864d90f99b744825a79a6daef2472b5d : SUCCESS in 3m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a9e934a7e11411f913d045a32efc959 : SUCCESS in 4m 38s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/a8be3a347ad1446994d492f258d5765c : FAILURE in 34m 12s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/34c11a20e30b440da219d03d974b274f : SUCCESS in 43m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d455dc313396fa924263f0461638a935cabb17ec","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-10 11:20:51.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a6e471f3ff9ac024e83fc26c3a68e2ccddcab6cd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-23 02:58:52.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"69b386eb7108d7bb50cf69836d57a4626464e19a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-05-23 03:00:34.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3606846ca05a84fa62ef8e2e7083460532896fd0","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-05-23 07:23:38.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9a99ddd520b34dd41579e5101b55fa4573bd6653","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-05-23 08:55:53.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"1b49c89c29ddb0addab1402e91f8127f5aae8c4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2023-06-01 06:44:16.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"eb6de3f4ab54104f34da6161d054edf3466aae06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 08:02:32.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170a706ac31d4894a9f5e33b0da2d385\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fc3f62dc89d4311bc5c16acae274335 : SUCCESS in 1h 17m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bb2b86d02f44143b3733f6aba2b16de : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/268c8913744e47cbb691b110b672b9a8 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b0f50fb660d489fb5367340270c25ec : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/542d9325e32e47d69f02ed3fb2bb55c6 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/056cc93abdec4789aa82d80fc017bf04 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b7e7e9ce0a4e83bf5b11e0727c43d1 : SUCCESS in 3m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecdaabb46879444da093f71cd3c0f16d : SUCCESS in 2m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c3a2365be9b04255ba3788e5cccf9f9e : SUCCESS in 3m 39s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/355142b196394e6a9b195da959743918 : SUCCESS in 42m 01s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/c6a09716fefa47a193a036cba46d64e1 : SUCCESS in 30m 28s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"252e80639d524ca1f30e1332bbed3d45c9efdef7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-06-01 13:35:03.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"cf6380fb0e6777485604b4672bc6fdb655315c53","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-06-06 08:19:34.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"da13effee092948b319f54be06c80045588012d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 08:20:03.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"c872bd775d5b03d199f7cfc1f6cecafde37a53f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 09:59:07.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c566fa10d2334913bcb1107110771ef4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/afbe1d99c88f477bb2745ba28c232a32 : SUCCESS in 1h 38m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff6d069ff80d4cc29bffd49376ddcb8d : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90213d8045df4bd78b13defd21fed0e4 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5120d353a38141a48d7746189877e2f4 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7ffaaf301a04475b2d0aa51ce599cd2 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a753f2ded95c4c58b7b841e7b4a2474d : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0123c7c0f48340d588f24a5e78d96d3c : SUCCESS in 2m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b7e993ad987647eda27e79d6524c94f9 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":12},{"id":"7af61376d0d200d76d0b4c2025b450086940a3a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 09:59:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"3ae10e6f7fcbb1ecb2f722e16a33e227fdc4fd35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 10:00:09.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b9dca6ed30d48be8baf3b209813a135\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57c4e7065aba4d0e802acbf52505757f : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/961da32f8c3643de955349e0fc0c2553 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"3c5feadda223ed8dd7d1dcce6c998ed48e0c6b7f","revisions":{"f9930897903a151e4d79f16e305dcb053eb9d3bf":{"kind":"REWORK","_number":1,"created":"2023-04-03 09:28:07.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/1"}}},"commit":{"parents":[{"commit":"4a4a13e7a996b081b70561e502e01cc221c60292","subject":"add support of jaeger+otlp for jaeger tracing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/4a4a13e7a996b081b70561e502e01cc221c60292"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-27 14:34:52.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-04-03 09:28:00.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/f9930897903a151e4d79f16e305dcb053eb9d3bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/f9930897903a151e4d79f16e305dcb053eb9d3bf"}]},"branch":"refs/heads/master"},"cfe1a5e1884bc5facfd4a7bdc463c33636517813":{"kind":"REWORK","_number":2,"created":"2023-04-04 08:21:15.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/2"}}},"commit":{"parents":[{"commit":"4a4a13e7a996b081b70561e502e01cc221c60292","subject":"add support of jaeger+otlp for jaeger tracing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/4a4a13e7a996b081b70561e502e01cc221c60292"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-27 14:34:52.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-04-04 08:21:06.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/cfe1a5e1884bc5facfd4a7bdc463c33636517813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/cfe1a5e1884bc5facfd4a7bdc463c33636517813"}]},"branch":"refs/heads/master"},"70f50721c6476af60c4a1aae45d15bba70f2e133":{"kind":"REWORK","_number":3,"created":"2023-04-04 09:43:20.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/3"}}},"commit":{"parents":[{"commit":"4a4a13e7a996b081b70561e502e01cc221c60292","subject":"add support of jaeger+otlp for jaeger tracing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/4a4a13e7a996b081b70561e502e01cc221c60292"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-27 14:34:52.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-04-04 09:43:11.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/70f50721c6476af60c4a1aae45d15bba70f2e133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/70f50721c6476af60c4a1aae45d15bba70f2e133"}]},"branch":"refs/heads/master"},"5f59735730ee745cd9462fe3002ef0d250941f56":{"kind":"REWORK","_number":4,"created":"2023-04-24 09:52:33.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/4"}}},"commit":{"parents":[{"commit":"040d297f7ef8e6ad473050c3c6adc20d1aa18272","subject":"add support of jaeger+otlp for jaeger tracing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/040d297f7ef8e6ad473050c3c6adc20d1aa18272"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-27 14:34:52.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-04-24 09:52:10.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/5f59735730ee745cd9462fe3002ef0d250941f56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/5f59735730ee745cd9462fe3002ef0d250941f56"}]},"branch":"refs/heads/master"},"a486f4fb72a3d124a744e6499b11fa7cfb476e14":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-04-24 13:53:34.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/5"}}},"commit":{"parents":[{"commit":"2bf0c23dc3910368695cd6efdf8a7cfb2da7087f","subject":"add support of jaeger+otlp for jaeger tracing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/2bf0c23dc3910368695cd6efdf8a7cfb2da7087f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-27 14:34:52.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-04-24 13:52:52.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a486f4fb72a3d124a744e6499b11fa7cfb476e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a486f4fb72a3d124a744e6499b11fa7cfb476e14"}]},"branch":"refs/heads/master"},"a4816b573c728c5ca58851469f524de28f05623b":{"kind":"NO_CHANGE","_number":6,"created":"2023-04-24 15:06:59.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/6"}}},"commit":{"parents":[{"commit":"675397d2460e7ca2ea179be33400bc452f21a1a3","subject":"add support of otlp exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/675397d2460e7ca2ea179be33400bc452f21a1a3"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-27 14:34:52.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-04-24 15:06:47.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a4816b573c728c5ca58851469f524de28f05623b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/a4816b573c728c5ca58851469f524de28f05623b"}]},"branch":"refs/heads/master"},"5f82bd54309155d8b7cfde31517af1b72838471a":{"kind":"REWORK","_number":7,"created":"2023-04-25 09:08:00.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/7"}}},"commit":{"parents":[{"commit":"675397d2460e7ca2ea179be33400bc452f21a1a3","subject":"add support of otlp exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/675397d2460e7ca2ea179be33400bc452f21a1a3"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-27 14:34:52.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-04-25 09:07:51.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/5f82bd54309155d8b7cfde31517af1b72838471a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/5f82bd54309155d8b7cfde31517af1b72838471a"}]},"branch":"refs/heads/master"},"4804ab0a6f13e616f872bdd3db4b208617caf465":{"kind":"REWORK","_number":8,"created":"2023-04-25 12:03:50.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/8"}}},"commit":{"parents":[{"commit":"675397d2460e7ca2ea179be33400bc452f21a1a3","subject":"add support of otlp exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/675397d2460e7ca2ea179be33400bc452f21a1a3"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-27 14:34:52.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-04-25 12:03:43.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/4804ab0a6f13e616f872bdd3db4b208617caf465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/4804ab0a6f13e616f872bdd3db4b208617caf465"}]},"branch":"refs/heads/master"},"67ab3a6bee9fabaab982eb39a2c640415c993058":{"kind":"REWORK","_number":9,"created":"2023-05-03 09:12:11.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/9"}}},"commit":{"parents":[{"commit":"6f0a46a0b559ba45513e9646a5b2a40d7ca74759","subject":"add support of otlp exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/6f0a46a0b559ba45513e9646a5b2a40d7ca74759"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-03-27 14:34:52.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-05-03 09:12:01.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/67ab3a6bee9fabaab982eb39a2c640415c993058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/67ab3a6bee9fabaab982eb39a2c640415c993058"}]},"branch":"refs/heads/master"},"4ee44ddabc32eca71df61ccf8e3c5620d9742697":{"kind":"REWORK","_number":10,"created":"2023-05-10 08:40:39.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/10"}}},"commit":{"parents":[{"commit":"908e7402320eb067db45aa9700d54d31c259f3ca","subject":"add support of otlp exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/908e7402320eb067db45aa9700d54d31c259f3ca"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-05-10 08:40:05.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-05-10 08:40:05.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/4ee44ddabc32eca71df61ccf8e3c5620d9742697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/4ee44ddabc32eca71df61ccf8e3c5620d9742697"}]},"branch":"refs/heads/master"},"196c9b53145e38a82ed2eead53b4d4aef3a8de66":{"kind":"REWORK","_number":11,"created":"2023-05-10 09:38:36.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/11"}}},"commit":{"parents":[{"commit":"908e7402320eb067db45aa9700d54d31c259f3ca","subject":"add support of otlp exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/908e7402320eb067db45aa9700d54d31c259f3ca"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-05-10 08:40:05.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-05-10 09:37:55.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/196c9b53145e38a82ed2eead53b4d4aef3a8de66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/196c9b53145e38a82ed2eead53b4d4aef3a8de66"}]},"branch":"refs/heads/master"},"3c5feadda223ed8dd7d1dcce6c998ed48e0c6b7f":{"kind":"REWORK","_number":12,"created":"2023-06-01 06:44:16.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/879318/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/18/879318/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/18/879318/12"}}},"commit":{"parents":[{"commit":"908e7402320eb067db45aa9700d54d31c259f3ca","subject":"add support of otlp exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/908e7402320eb067db45aa9700d54d31c259f3ca"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-05-10 08:40:05.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2023-06-01 06:44:08.000000000","tz":120},"subject":"profiler: add python requests profile","message":"profiler: add python requests profile\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I00e9f5661a3bd54acc846e8c326896d21e70be36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/3c5feadda223ed8dd7d1dcce6c998ed48e0c6b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/3c5feadda223ed8dd7d1dcce6c998ed48e0c6b7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
