)]}'
{"id":"openstack%2Fosprofiler~900797","triplet_id":"openstack%2Fosprofiler~master~If4492943199c75c294e4eb5a3600cf5d2bdad0b8","project":"openstack/osprofiler","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-11-14 12:13:21.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-16 17:50:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If4492943199c75c294e4eb5a3600cf5d2bdad0b8","subject":"List up extra requirements for drivers","status":"MERGED","created":"2023-11-13 13:28:10.000000000","updated":"2023-11-16 17:51:40.000000000","submitted":"2023-11-16 17:50:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900797","meta_rev_id":"679314efc7b3294c69b73f32a8dcae9d4d884934","_number":900797,"virtual_id_number":900797,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 17:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-11-14 12:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-11-16 15:54:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-13 13:32:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-11-13 15:21:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2a399830bf9d859be6d887dd79b9c508a6ebc33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-13 13:28:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20f8966ee0f861981d9e607cbfe6aa148ec9fcdf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-13 13:32:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be906d4efef12782d9acdd77046be88029b83f36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-13 13:37:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"6e148ae82172b85ef4739382d07a5c6732d8a3a5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-13 13:37:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8aefdf225145e563c815feb2ae71ef6d9b038e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 15:21:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/503d771d57c8410eaf98e8cd5d6d8a17\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2d77bfc5e0b34a5a89d31ada6e7a87dd : SUCCESS in 2m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/317a68540abb4e49a9c9f630eb8bc538 : FAILURE in 2m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3bbb34de586b4074a80a5935dca9e25c : SUCCESS in 1h 36m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2c03568c07a43428f33d0c5aa63fe63 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a99aae0aba248f78d98aeec64c4eb7e : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c031370eb6747a38fc3af8f259c74fb : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc20b084dd114cb4bcf8c74e96d5063d : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/011e84d7ae2a4f688229a82892e5525a : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd58cc85d1d245aeb89121097514c2b3 : SUCCESS in 2m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2a17927fecf64f31878159ee07b31c6b : SUCCESS in 3m 59s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/b18c984e12f646dab14109a418d1dd85 : POST_FAILURE in 27m 39s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/d514c3afd6ec48e5961fa3ecf274cdad : POST_FAILURE in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"51a3245e444c8321033f37212334fb2b31dcca68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-13 15:31:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6d1a9d9b05e3e8866336fd49b84d0f137e57d008","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-13 15:40:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a5c9faac5f3590a3e9b1c0df602cdb7e8e0b0322","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-13 16:33:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"405e75b2b5237523047f4f1d27e1a78daee3555c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 16:59:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d8f1d668aaa4213afed6b276aa97198\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2b6815c5280e40af921b2ffa43be6c88 : SUCCESS in 2m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/443112b495214c32af0a53ba276d24e9 : FAILURE in 3m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3308ec1c20a24897b6a8cc069e47479f : SUCCESS in 1h 18m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e14e1cf85c514f21a854f1a632e81594 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9455ca3511ee49b68b346bf8ad30b6d0 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2da8b092ed3148948b00410c6f6a5814 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/961dd19be2dd4f3b9d7b851209e96f39 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7632be4e47324dc0baafb73be9cfe0d0 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/598444e6e0ef4f06b0c50843d161396c : SUCCESS in 3m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1de63a26f00d4c36a7a8c7f7340fb89f : SUCCESS in 4m 03s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/7a22fa3e0a644b338d02b026af88a628 : POST_FAILURE in 37m 54s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/9d881dc7efeb4133bdc61abf1fdf41ad : POST_FAILURE in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fef53754c0be9f2e1c161286ca2ab5d505f48366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-14 02:07:02.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2156ba5e89e17e535b4c95d6f56f12cec0d2aa34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 04:43:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dab422d5d227491482c7971532dd0fae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0ada05e8f1d74e65adee6ff48242fb59 : SUCCESS in 2m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/51c80a1c49c34bc8a4c1190df2834f67 : SUCCESS in 2m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02de486ff82e41eaacd59a3d061b62ab : SUCCESS in 1h 30m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbb9755d7797442e8dbbace87e2e0380 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ae4c11d6214f71ab8c15ce65329909 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/141283ed2a684c99b38090462429d024 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b616e996b6d4ff9b3d39cfe29b5b3b1 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62bfad8fdfa54d81927a8235fbacfe97 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbc087c7da7a4df9bc4a90716367dd07 : SUCCESS in 3m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/945b6328085144da83aac60f6cbd6769 : SUCCESS in 4m 04s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/3099be989b634f3ea2277c67f4b9bfa9 : POST_FAILURE in 35m 00s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/91e782d046b14f949a3775f7319ee0ff : POST_FAILURE in 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dcff2d5574d9a4fcfc5323c9d18f1675f9e58c0a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-14 12:13:21.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"40684d9d3025c5ae72978e5979fe8321e9e8be04","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-16 15:54:15.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1b521baebe95267be11ab72b3182204041a0f5f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 15:54:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4e58da6bc01456a76f056191220fda5c2e3a612f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 17:50:31.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/792b1aa6236142ef96452a076d283bfa\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/83c06702647b4a12b2b1fb23e1b6fb94 : SUCCESS in 2m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0627c19ce1f9469880d7de30c663bbda : SUCCESS in 3m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b956708d03c84af4ace08115df858c0a : SUCCESS in 1h 50m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d789eb630588416d824156d441015e09 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/011e456d5d064dfdb28180b948a18450 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4b846b50fac461da87dce283d369266 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e54f063d64745ae92614f607c27721e : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db26e12995ba4489a0e29d75b02a8991 : SUCCESS in 3m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/68f2a8dca4a74b7584777321c17abb2d : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":5},{"id":"80a6cdd2aac5ac707a50bf0c6001ded245f6699d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 17:50:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"679314efc7b3294c69b73f32a8dcae9d4d884934","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 17:51:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f7d381095de4d32bc62c9b3c4591ded\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b5b2211f18c47a4af77a90364382977 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/665761e877db4e749bfb4053d75732cc : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"17835cb3dff310a874c486f95846174e4aeb2f30","revisions":{"c37ee083efa1563fea9c6d610fbd002038a80761":{"kind":"REWORK","_number":1,"created":"2023-11-13 13:28:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/900797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/97/900797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/1"}}},"commit":{"parents":[{"commit":"ac796bdabf854356bcbae8b4aafaf36e5659f6a5","subject":"Move oslo.config to normal requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/ac796bdabf854356bcbae8b4aafaf36e5659f6a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-13 13:27:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-13 13:27:06.000000000","tz":540},"subject":"List up extra requirements for drivers","message":"List up extra requirements for drivers\n\nMost of the drivers require extra requirements. This adds these to\nthe extras field in setup.cfg so that users can more easily install\nextra requirements.\n\nChange-Id: If4492943199c75c294e4eb5a3600cf5d2bdad0b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/c37ee083efa1563fea9c6d610fbd002038a80761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/c37ee083efa1563fea9c6d610fbd002038a80761"}]},"branch":"refs/heads/master"},"6aece980aaea3e517a3a20a1ec4453c39edf0d20":{"kind":"REWORK","_number":2,"created":"2023-11-13 13:37:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/900797/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/97/900797/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/2"}}},"commit":{"parents":[{"commit":"ac796bdabf854356bcbae8b4aafaf36e5659f6a5","subject":"Move oslo.config to normal requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/ac796bdabf854356bcbae8b4aafaf36e5659f6a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-13 13:27:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-13 13:36:52.000000000","tz":540},"subject":"List up extra requirements for drivers","message":"List up extra requirements for drivers\n\nMost of the drivers require extra requirements. This adds these to\nthe extras field in setup.cfg so that users can more easily install\nextra requirements.\n\nChange-Id: If4492943199c75c294e4eb5a3600cf5d2bdad0b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/6aece980aaea3e517a3a20a1ec4453c39edf0d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/6aece980aaea3e517a3a20a1ec4453c39edf0d20"}]},"branch":"refs/heads/master"},"35297294f0b43962df5ba17f6ac5fe570dc38c71":{"kind":"REWORK","_number":3,"created":"2023-11-13 15:31:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/900797/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/97/900797/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/3"}}},"commit":{"parents":[{"commit":"ac796bdabf854356bcbae8b4aafaf36e5659f6a5","subject":"Move oslo.config to normal requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/ac796bdabf854356bcbae8b4aafaf36e5659f6a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-13 13:27:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-13 15:31:25.000000000","tz":540},"subject":"List up extra requirements for drivers","message":"List up extra requirements for drivers\n\nMost of the drivers require extra requirements. This adds these to\nthe extras field in setup.cfg so that users can more easily install\nextra requirements.\n\nChange-Id: If4492943199c75c294e4eb5a3600cf5d2bdad0b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/35297294f0b43962df5ba17f6ac5fe570dc38c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/35297294f0b43962df5ba17f6ac5fe570dc38c71"}]},"branch":"refs/heads/master"},"146fd65c5007400885e20fc6f766151cf9b374ab":{"kind":"REWORK","_number":4,"created":"2023-11-13 15:40:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/900797/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/97/900797/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/4"}}},"commit":{"parents":[{"commit":"ac796bdabf854356bcbae8b4aafaf36e5659f6a5","subject":"Move oslo.config to normal requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/ac796bdabf854356bcbae8b4aafaf36e5659f6a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-13 13:27:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-13 15:40:23.000000000","tz":540},"subject":"List up extra requirements for drivers","message":"List up extra requirements for drivers\n\nMost of the drivers require extra requirements. This adds these to\nthe extras field in setup.cfg so that users can more easily install\nextra requirements.\n\nAlso removes the lower-constraints.txt file because we already removed\ntests of l-c by [1].\n\n[1] bcca5bd4a8c3fed96dfcf01bf24233fa509e610b\n\nChange-Id: If4492943199c75c294e4eb5a3600cf5d2bdad0b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/146fd65c5007400885e20fc6f766151cf9b374ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/146fd65c5007400885e20fc6f766151cf9b374ab"}]},"branch":"refs/heads/master"},"17835cb3dff310a874c486f95846174e4aeb2f30":{"kind":"REWORK","_number":5,"created":"2023-11-14 02:07:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/97/900797/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/97/900797/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/97/900797/5"}}},"commit":{"parents":[{"commit":"ac796bdabf854356bcbae8b4aafaf36e5659f6a5","subject":"Move oslo.config to normal requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/ac796bdabf854356bcbae8b4aafaf36e5659f6a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-13 13:27:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 02:05:47.000000000","tz":540},"subject":"List up extra requirements for drivers","message":"List up extra requirements for drivers\n\nMost of the drivers require extra requirements. This adds these to\nthe extras field in setup.cfg so that users can more easily install\nextra requirements.\n\nAlso removes the lower-constraints.txt file because we already removed\ntests of l-c by [1].\n\nNote that the lower versions of SQLAlchemy and oslo.messaging are\npicked up from requirements in the other repos, and further lower\nversions might be allowed.\n\n[1] bcca5bd4a8c3fed96dfcf01bf24233fa509e610b\n\nChange-Id: If4492943199c75c294e4eb5a3600cf5d2bdad0b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/17835cb3dff310a874c486f95846174e4aeb2f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/17835cb3dff310a874c486f95846174e4aeb2f30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
