)]}'
{"id":"openstack%2Fosprofiler~932480","triplet_id":"openstack%2Fosprofiler~master~I316119d1b9a3e8aad75c764e5a6d64a6e7069bcb","project":"openstack/osprofiler","branch":"master","topic":"remove-py38","hashtags":[],"change_id":"I316119d1b9a3e8aad75c764e5a6d64a6e7069bcb","subject":"Remove Python 3.8 support","status":"MERGED","created":"2024-10-16 02:46:18.000000000","updated":"2024-11-15 16:38:27.000000000","submitted":"2024-11-15 16:35:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932480-remove-py38","meta_rev_id":"40e3e1dcef193ee6226a59219422cc4941952b69","_number":932480,"virtual_id_number":932480,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 16:35:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2024-11-15 15:08:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-11-15 15:08:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-16 04:47:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26c3d99425c8c723f147826b29ec8e7202de13df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-16 02:46:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d01717af142b4fecb312bc9fa6d27edcaaa4121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 04:47:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d18b9c2f6102445c89d53f9f3348ffd5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d38ec55cc7c1482bb7dbae88fe044b2d : SUCCESS in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b69d1d86065644ce887d1d16433e726f : SUCCESS in 1h 50m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49e7a3077ff340aba728cd7dab12b194 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b53987dfc9de4b969b3fd20cf0bc4ce8 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32a69ba159884489954113bc5f3c3050 : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a751491de8c4c5d8a9b3a3eebf0a536 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ff1a2ca6f9c4e3f85e1be856167d316 : SUCCESS in 3m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d650d3a07194fc187edf1dba2156332 : SUCCESS in 2m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/701082ece0994017a80309b8b1e5f39f : SUCCESS in 3m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/47de0830565a479eae972811637eebd7 : SUCCESS in 7m 05s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/44443b2875884139b3f38872e5803eb0 : POST_FAILURE in 32m 11s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/4e28864f6fbb4025a93f19a8fdf3dc4a : POST_FAILURE in 44m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15b21164ab4d5545ffed2abcc210032203bfc136","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-16 13:37:46.000000000","message":"Topic set to remove-py38","accounts_in_message":[],"_revision_number":1},{"id":"4961bd73afeceb895b071066d156dcacc5f4d716","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-15 15:08:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ccd35302b1ac61d1288810b38b18b13ae76ecd37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:08:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d0783f2b63acd40b159e310d5cc60ea477aa71d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:35:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/546e93c1207343199e43c96436adf428\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/acbe0fa736d24ffd986c55fa7f08fcf8 : SUCCESS in 1m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aed2c23a96654768bbc884c7861e0317 : SUCCESS in 1h 23m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72231893b4904acc9149de087139c948 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/544c8f164d414e54bbbdb3a420dac8a3 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dff773fc22fa46529cfb1dc25588596e : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72d4382fb28f4a2db9fde24b0a84c904 : SUCCESS in 2m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/724ff1327bd94d8494aca86c244df28b : SUCCESS in 2m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/16f72e82070e4e13b9ec9f9550cf2eb3 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5282ba2feee817580665cb121c39ab9b35cc4772","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:35:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"40e3e1dcef193ee6226a59219422cc4941952b69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 16:38:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3e00da827b945d5b7960bea46c14468\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a3e831c55d0458c94d78d6ee9ac71fb : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92c9df332bb241388e374cfb271f46f1 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f9847298b95d26e8ec226accc7bcbcbe872383b","revisions":{"0f9847298b95d26e8ec226accc7bcbcbe872383b":{"kind":"REWORK","_number":1,"created":"2024-10-16 02:46:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/80/932480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/80/932480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/80/932480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/80/932480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/80/932480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/80/932480/1"}}},"commit":{"parents":[{"commit":"92dee3ed9e9f45f6a31573606fbb0cc75bdea2e0","subject":"Merge \"Unpin for newer elasticsearch versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/92dee3ed9e9f45f6a31573606fbb0cc75bdea2e0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-16 02:45:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-16 02:45:21.000000000","tz":540},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 is no longer part of the tested runtimes for 2024.2[1]\nbecause its EOL is coming soon.\n\nAlso update the python versions in functional jobs to reflect\nthe supported versions.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I316119d1b9a3e8aad75c764e5a6d64a6e7069bcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0f9847298b95d26e8ec226accc7bcbcbe872383b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/0f9847298b95d26e8ec226accc7bcbcbe872383b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
