)]}'
{"id":"openstack%2Fosprofiler~935594","triplet_id":"openstack%2Fosprofiler~master~Ie142e4c50e7cec0a6364f22b6781edc7186e95bd","project":"openstack/osprofiler","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-01 10:27:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie142e4c50e7cec0a6364f22b6781edc7186e95bd","subject":"Drop redundant if-else block","status":"MERGED","created":"2024-11-19 02:59:56.000000000","updated":"2025-04-01 10:28:48.000000000","submitted":"2025-04-01 10:27:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935594","meta_rev_id":"da19a8aa5636a6fca438bbc8b97607e63331b26e","_number":935594,"virtual_id_number":935594,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-01 10:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-04-01 09:21:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-04-01 09:21:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 04:23:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-01 09:21:17.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"8bb4cba0938b4147d689ae6609b0affaa14afb6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 02:59:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94443885a2195b6f95c5bafeb399688303727788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 04:23:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48d2fe9450dd4a83a37a2bce6412fcea\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/455cd83929e5433bbcf3e7be9d35e154 : SUCCESS in 1h 21m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ab1eb6d24f24f048f95aeadeb7c3f72 : FAILURE in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f28cc9f48152494380cb386d8c9d40f5 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cca4e4119d7a4e3e86c6fef82210d62d : FAILURE in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8bf896e8d5b342b493bfd40ba5439ab5 : FAILURE in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e01aadbc32443a1be0827f38e748bde : SUCCESS in 3m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f1f9442b2a6c4a53acf2c92f7aac17eb : SUCCESS in 5m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9142a8b382774a27841ac61f28d4eef9 : FAILURE in 6m 03s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/f5cba10971cc4b6bbd694baadd2be80c : POST_FAILURE in 25m 50s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/987cb6b16c9248b8be82d7c031b3d1c0 : POST_FAILURE in 27m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff28fc702e9b6b537b5ed65e3257573d4191a5a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-19 14:08:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"058467a67c3f19b2f1fd8fbbae676bf73da80202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 16:04:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6815da0a7c604ec396bfd55a9cee5aeb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9daae65031d40fd9322f34db6c77ac3 : SUCCESS in 1h 47m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa122953754c480da069c346b5d6b8ef : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7546cb7d27b04b968758640d966c160f : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd7b5dbf046e4188a3b1a712ab2d2fec : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d90a7dd6bbf46d281d3ccef4c84b8d3 : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b52c2f628524248be101deb5dd143ac : SUCCESS in 3m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f6e965fd09bb4e6ab1e70ff1f6ab0816 : SUCCESS in 3m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/680fa563902a43ed9b45f2d9ef502067 : SUCCESS in 8m 24s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/ecccecd729f040d09c056066569e9e02 : POST_FAILURE in 40m 16s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/2242c63573df4657a5c6d13a078b7153 : POST_FAILURE in 39m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09bf3b0ff0e68794287e5356eacb1b55d1181e0e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-04-01 09:21:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"17c8383d90d5be4c90b99d7ebe233d9ea87cc9fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 09:21:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d74497f6fe057f8eb8b966800f8e8ca03106d86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 10:27:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5db1ead4fde74d83ba7121a2378a85c5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/040c88e9b7a5499dbd87cc3612c0846b : SUCCESS in 1h 02m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f64b9eb81f1417fb119766c20ed66da : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7223148566f9469a901026cf14d4c643 : SUCCESS in 3m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff98efe032484f26a86ac27274cd92f4 : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08df61282bd3447fa0c58bdd110b83ce : SUCCESS in 3m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d846866c48e74743aa978110260fe512 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":2},{"id":"3b7298869ef630e70ead6959e0745ae4ac6c27ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 10:27:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"da19a8aa5636a6fca438bbc8b97607e63331b26e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 10:28:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcb18696d64b402cba909d29c68182ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25ded53484454e5cb799e0237275c6d1 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d782c776ed2a5e8cb9b3d3be246ea13d34536d2","revisions":{"5339d4f8470e411fcc1af837d9fdaa6bc1dde233":{"kind":"REWORK","_number":1,"created":"2024-11-19 02:59:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/935594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/94/935594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/94/935594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/94/935594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/94/935594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/94/935594/1"}}},"commit":{"parents":[{"commit":"309c0798f954f0062fa028826775f7a3b8465489","subject":"Merge \"Replace deprecated constant_time_compare\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/309c0798f954f0062fa028826775f7a3b8465489"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-19 02:59:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-19 02:59:29.000000000","tz":540},"subject":"Drop redundant if-else block","message":"Drop redundant if-else block\n\nChange-Id: Ie142e4c50e7cec0a6364f22b6781edc7186e95bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/5339d4f8470e411fcc1af837d9fdaa6bc1dde233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/5339d4f8470e411fcc1af837d9fdaa6bc1dde233"}]},"branch":"refs/heads/master"},"8d782c776ed2a5e8cb9b3d3be246ea13d34536d2":{"kind":"REWORK","_number":2,"created":"2024-11-19 14:08:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/935594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/94/935594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/94/935594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/94/935594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/94/935594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/94/935594/2"}}},"commit":{"parents":[{"commit":"309c0798f954f0062fa028826775f7a3b8465489","subject":"Merge \"Replace deprecated constant_time_compare\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/309c0798f954f0062fa028826775f7a3b8465489"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-19 02:59:29.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-19 14:07:57.000000000","tz":540},"subject":"Drop redundant if-else block","message":"Drop redundant if-else block\n\nThe info argument defaults to None, so call with explicit info\u003dNone\nresults in the same behavior as the call without any argument and this\ncan be leveraged to simplify the logic.\n\nChange-Id: Ie142e4c50e7cec0a6364f22b6781edc7186e95bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/8d782c776ed2a5e8cb9b3d3be246ea13d34536d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/8d782c776ed2a5e8cb9b3d3be246ea13d34536d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
