)]}'
{"id":"openstack%2Fosprofiler~961011","triplet_id":"openstack%2Fosprofiler~master~I04233fb985f29cb7c1278042835e09a93f570ace","project":"openstack/osprofiler","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-15 12:13:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I04233fb985f29cb7c1278042835e09a93f570ace","subject":"Replace own hacking checks by built-in ones","status":"MERGED","created":"2025-09-13 19:14:21.000000000","updated":"2025-09-15 12:14:01.000000000","submitted":"2025-09-15 12:13:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"961011","meta_rev_id":"428e78906ca668bdf7f9dc4d804c78b7369205cd","_number":961011,"virtual_id_number":961011,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-15 12:13:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-09-15 10:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-09-15 10:35:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-13 21:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-15 10:35:38.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"7c32a6971e7b996fed02d287ac82fb975e7cc6a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-13 19:14:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56a66d6d97e7343f7834ac68ada79f9b0927d455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-13 21:07:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b24d228e7574983a0e6d016bf52f9ac\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b331cdb94afe4ffda6b1e42d2b37ab57 : SUCCESS in 1h 48m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/291d928a8b584c68bcabddc8277e6c6f : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4a983d919da46beb78a93148b58226d : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40a14c67787d4e73b139b1e6499ecebb : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/127deaff3bf5478293d7394081d9fb44 : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b45db2f496834f2bab8ead849500c118 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/204bd746bb604d9797d3c1c78b69cdec : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c07c58b161884c5da87a1a869b7100ed : SUCCESS in 2m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc9ab0d53c164cefa8c9c6eda79dcb58 : SUCCESS in 3m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/967a4283b5d94605b60851c1293e4962 : SUCCESS in 6m 28s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/31ed581984744576b2f4e479e5203158 : POST_FAILURE in 36m 44s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/17dafb24ba6d4946be260b54685327ae : POST_FAILURE in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"193a0487eea1bec7cb4807a222621e1517bbf077","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-09-15 10:35:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"59d203185fc656c4d5ecca34866bbd81b91a0e44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 10:36:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91d81a2b83fec9104fccc333f56a9e9b2b7d6240","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 12:13:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f762b301da7f404a96e7fb8d7de40a1f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f56e0bba521a4f5e81299e4805988976 : SUCCESS in 1h 35m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/affc69147348422da3127ec2b81f1e95 : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c4128955abe4650b0db0e9d7667bc28 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7db138947902420893b9053b745a54d3 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8610d26676c485bb75c0c0d0d5f4dd2 : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/993031f8006649c4ad349da0037b9249 : SUCCESS in 3m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/89bc8b25c0a1471397ed35dbc166ea87 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"15d2a89a204ecfa5954ec41fd8faba92e0d7b46a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 12:13:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"428e78906ca668bdf7f9dc4d804c78b7369205cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 12:14:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/491f12866ceb48969b4418bb809a66b1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a33fae5617b4e35a4d33f2335a5edf6 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91b53f4191154527aa92c99cdaee62ff : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6839afe26c2a749c8ed2e4b4f931c80b9cd4fe2","revisions":{"c6839afe26c2a749c8ed2e4b4f931c80b9cd4fe2":{"kind":"REWORK","_number":1,"created":"2025-09-13 19:14:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/961011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/11/961011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/961011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/961011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/11/961011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/11/961011/1"}}},"commit":{"parents":[{"commit":"ff2c6c43c02be0a51efa1e1a3ff0a72c1c508865","subject":"Merge \"Drop Python 3.9 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/ff2c6c43c02be0a51efa1e1a3ff0a72c1c508865"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 19:12:51.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 19:14:08.000000000","tz":540},"subject":"Replace own hacking checks by built-in ones","message":"Replace own hacking checks by built-in ones\n\nRecent hacking provides built-in checks to reject specific usage of\nassertions. Replace the own hacking checks by built-in ones to\nleverage the shared implementation.\n\nNote that hacking_assert_is_none is enabled by default, so it\u0027s not\nexplicitly enabled.\n\nChange-Id: I04233fb985f29cb7c1278042835e09a93f570ace\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/c6839afe26c2a749c8ed2e4b4f931c80b9cd4fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/c6839afe26c2a749c8ed2e4b4f931c80b9cd4fe2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
