)]}'
{"id":"openstack%2Fosprofiler~979500","triplet_id":"openstack%2Fosprofiler~master~I71f003998461723241edd7fc16f8a7970ebdc0d0","project":"openstack/osprofiler","branch":"master","topic":"typing","hashtags":[],"change_id":"I71f003998461723241edd7fc16f8a7970ebdc0d0","subject":"Add ruff","status":"MERGED","created":"2026-03-09 11:32:54.000000000","updated":"2026-03-09 15:30:07.000000000","submitted":"2026-03-09 15:28:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979500-typing","meta_rev_id":"7287b1a3b93e80cb53cd2b39768c043f7acbd20e","_number":979500,"virtual_id_number":979500,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-09 15:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-09 13:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-09 13:38:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 13:40:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"668a3ea1811e3aa5831605b14f48bc93a2388966","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 11:32:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea4c76120b8d0e3dadc988711524ff75f9fdc306","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 13:38:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1dd64690f17def9f56c809ed629792222509f948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:40:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7b1fa9d344f4404ada701bef800f449\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c6af35a6fd914710a6a490e7ee8ae143 : SUCCESS in 3m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6d023f92abc740beb608b6fc7097ddc0 : SUCCESS in 3m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/21fcd0e1253c4866a438ab3714257805 : SUCCESS in 1h 41m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc099622f0724a7591dd28ba29fecb3f : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c59c20db701b4c309af8c4cb01518c52 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6c8721a6bed48e7b2615cee4eaf24be : SUCCESS in 3m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/24bc10c305d64661b852c84f1f0d7d9e : SUCCESS in 4m 03s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/155d70a94e5848d7a841c7484addd970 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db3af66d083f4fe4af7ff9df38bc5230 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e974af4e3104b5aaba6a063ad2ab7ca : SUCCESS in 3m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/919c9591a6a846909a9e048815eb7d8d : SUCCESS in 7m 53s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4fded1941c5449688cd3f30d001036f7 : SUCCESS in 4m 12s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/6a434f9de439487abbcf684b41fbcdaf : POST_FAILURE in 37m 53s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/5d63cf61375445fc8b9676beac99e2ac : POST_FAILURE in 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33ae4106ec050cc1bfa139e5cb3a5c5c50db674b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:41:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d53c94f8bd5e9fcdf7fad0d58d564c260ea5831f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:28:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/238ac59e124f45ec924b1204c0434ab5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/add03ed8966f447883c668eb2212a9dc : SUCCESS in 1m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f6cfa318033453382a4c97e603c3b0a : SUCCESS in 2m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ecdc714e80e474ba0f091c071cfeae4 : SUCCESS in 1h 42m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e891c85ba8814ac58bd2bb6b8eacca6f : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2effa0d45134a2c8d3ca155b6ddf346 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5df87b22e50b40159c4c083b3a2834b8 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f482211bfb254db8a4ec6a2f385272f5 : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9df156f067c94155a301706d44f8a005 : SUCCESS in 2m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f213221fd8043fa8dcd51fea251320d : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fbe23208a973a07befeb06d7543af54077a3ebdf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:28:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7287b1a3b93e80cb53cd2b39768c043f7acbd20e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 15:30:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77032d3f0fc44c70b65abccdddafd8c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56952bc1a694a07839e0bdde57d7c1a : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa448e07ca314aedb2f7456eb669dcf1 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bd377bc3c2cc3c61c1a6c0541e73c30019803c50","revisions":{"bd377bc3c2cc3c61c1a6c0541e73c30019803c50":{"kind":"REWORK","_number":1,"created":"2026-03-09 11:32:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/00/979500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/00/979500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/979500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/979500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/00/979500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/00/979500/1"}}},"commit":{"parents":[{"commit":"5a01ecb559a6b344be9ad7f447791e7c99040c56","subject":"Merge \"Fix KeyError in notifying db errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/5a01ecb559a6b344be9ad7f447791e7c99040c56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-08 16:06:20.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-08 18:49:01.000000000","tz":0},"subject":"Add ruff","message":"Add ruff\n\nChange-Id: I71f003998461723241edd7fc16f8a7970ebdc0d0\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/bd377bc3c2cc3c61c1a6c0541e73c30019803c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/bd377bc3c2cc3c61c1a6c0541e73c30019803c50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
