)]}'
{"id":"openstack%2Fosprofiler~979502","triplet_id":"openstack%2Fosprofiler~master~I2d32ba5bd94ffb5ccec29a4277894c911df17c07","project":"openstack/osprofiler","branch":"master","topic":"linter-sync","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-09 14:13:53.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I2d32ba5bd94ffb5ccec29a4277894c911df17c07","subject":"Update packaging configuration","status":"MERGED","created":"2026-03-09 11:33:21.000000000","updated":"2026-05-02 16:05:30.000000000","submitted":"2026-05-02 16:04:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979502-linter-sync","meta_rev_id":"14173dcc1677d515b3400fad9731e7dd6565b699","_number":979502,"virtual_id_number":979502,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-02 16:04:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-05-02 12:27:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-05-02 12:27:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-09 13:30:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8363a4560bd24203b5e8a861018f1ce81fb3063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 11:33:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee7998bd6d3b6377e253f17f703946faaf0fc787","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 11:34:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9670eed2a6960d0fd62cf2838009e82b5f0328fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-09 13:30:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/daa8716e80394013aed3142ffdeea6d8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/47c1a0a9da434803a5f73a4c3ef0f3ce : FAILURE in 2m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/58a5bcfc3db14185bbc8ccffb84876d7 : SUCCESS in 3m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b04f840508cf40a0bcfa338bc5a7ddf7 : SUCCESS in 1h 27m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e6258e909b94f48ac9c90634b126fcb : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/754742f42384497b8b29cde74a4e754f : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a507f048d2e147a394b2f1423f6b4249 : SUCCESS in 3m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cb04de7d52b4927944af8a04c9e8b20 : SUCCESS in 4m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/37744987f87e48d895a8af094c441747 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f50f589effb5460ca65a62e14bd4e5a6 : SUCCESS in 4m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a199a7517af4494ca1cac0dcd699dc07 : SUCCESS in 3m 59s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/661eb0c3a38b4283afc84c72a47a03e6 : SUCCESS in 4m 02s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/7f8778eac4a440ba851b93bb0633c4cf : POST_FAILURE in 21m 47s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/a2bf17914cde49988d8eea568f062dd1 : POST_FAILURE in 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac6f20a9bc723f937f4ecc66b568e8c48f54839a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-09 14:13:53.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"93ba4f25c29b08aa85209733225a9d9cb399ccd8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-02 12:27:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d790e6724ba2105a811832497b5f9e203452e00e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 14:18:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed08bc2533f7464082d8dc7e220a85af\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c07ae27487d749a1854a4c69ddfed677 : SUCCESS in 1m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/98910156885d46538ba54e8576593be9 : SUCCESS in 1m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45b9a05443eb473d958ee22a09085e82 : SUCCESS in 1h 41m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0703c485c884859927bc30e98ec2814 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b8b3fe6eb70454d9785606d4505c0c4 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e44e3e2e27f749b9b160fcee8a85fddd : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/92675b5484d9427aae4dda97f6ebc837 : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/74366f2ca07f459f93ace3510cd5c6f2 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c36d92d330d471ebcc259151228c181 : SUCCESS in 3m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01ffffea6c384e30aba103aacd712500 : SUCCESS in 3m 13s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/bf8100ebce164318ad5ce4a11ca846a0 : SUCCESS in 3m 15s\n- tempest-smoke-py3-osprofiler-redis https://zuul.opendev.org/t/openstack/build/5b42d5cf44a04ddc917e9f2be52a82d7 : POST_FAILURE in 33m 55s (non-voting)\n- tempest-smoke-py3-osprofiler-sqlalchemy https://zuul.opendev.org/t/openstack/build/2dee9407e35545f68c418f67a40f7e06 : POST_FAILURE in 42m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59ec0f7e7c912740215176e35dbd0c1a44e07e18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 14:19:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18c45402a643b26c848fc5636de81928b159a88a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 16:04:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0051bd85cfaf4dd390cb10a71566a2d5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2d0cb5a3ceee4f46947f8b0c987ee724 : SUCCESS in 2m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3284f82fc36649a1871a7c7e78ee7aba : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ba4924e2ec6f4103a7f1a0886bbd6f14 : SUCCESS in 1h 40m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eebfc17454f24c2aa324c4766f1f253b : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4df63c5a544a457e8f2c0c5519b2ec0d : SUCCESS in 2m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e2581b80bf7417e99b9d3bc47684c04 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/934ad727c3414945b228f8e8213ad13c : SUCCESS in 3m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0338707367e1435691cf5b629d0d502a : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f4d1347d87e90c8a728e483cf0c461f445d58126","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 16:04:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"14173dcc1677d515b3400fad9731e7dd6565b699","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-02 16:05:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ba94e7243f4c02b182dc863fe0175f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8f6ef75492746d281bff00423da3721 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"11848dc4d625dca5e4777e53de3d1535e0cc05f7","revisions":{"80d9d2130cfaadedb6e03a22f09ae207d0fabe94":{"kind":"REWORK","_number":1,"created":"2026-03-09 11:33:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/02/979502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/02/979502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/02/979502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/02/979502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/02/979502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/02/979502/1"}}},"commit":{"parents":[{"commit":"5a01ecb559a6b344be9ad7f447791e7c99040c56","subject":"Merge \"Fix KeyError in notifying db errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/5a01ecb559a6b344be9ad7f447791e7c99040c56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 11:33:13.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 11:33:13.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nResolve setuptools deprecation warnings around license configuration.\n\nChange-Id: I2d32ba5bd94ffb5ccec29a4277894c911df17c07\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/80d9d2130cfaadedb6e03a22f09ae207d0fabe94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/80d9d2130cfaadedb6e03a22f09ae207d0fabe94"}]},"branch":"refs/heads/master"},"11848dc4d625dca5e4777e53de3d1535e0cc05f7":{"kind":"REWORK","_number":2,"created":"2026-03-09 11:34:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/02/979502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/osprofiler","ref":"refs/changes/02/979502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/02/979502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/02/979502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/osprofiler refs/changes/02/979502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/osprofiler refs/changes/02/979502/2"}}},"commit":{"parents":[{"commit":"5a01ecb559a6b344be9ad7f447791e7c99040c56","subject":"Merge \"Fix KeyError in notifying db errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/5a01ecb559a6b344be9ad7f447791e7c99040c56"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 11:33:13.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-03-09 11:34:24.000000000","tz":0},"subject":"Update packaging configuration","message":"Update packaging configuration\n\nEnable package discovery and resolve setuptools deprecation warnings\naround license configuration.\n\nChange-Id: I2d32ba5bd94ffb5ccec29a4277894c911df17c07\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/11848dc4d625dca5e4777e53de3d1535e0cc05f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/osprofiler/commit/11848dc4d625dca5e4777e53de3d1535e0cc05f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
