)]}'
{"id":"openstack%2Fossa~1006088","triplet_id":"openstack%2Fossa~master~Ic0667090d0a7e0a117aea09bbf42b54474889162","project":"openstack/ossa","branch":"master","topic":"ossa-2026-040","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-09-21 17:04:48.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2026-09-21 17:04:48.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2026-09-21 09:47:34.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":["CVE-2026-93852","CVE-2026-93854","OSSA-2026-040"],"change_id":"Ic0667090d0a7e0a117aea09bbf42b54474889162","subject":"Add OSSA-2026-040 (CVE-2026-93852, CVE-2026-93854)","status":"MERGED","created":"2026-09-17 17:18:39.000000000","updated":"2026-09-21 17:06:44.000000000","submitted":"2026-09-21 17:04:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006088-ossa-2026-040","meta_rev_id":"2c2e9f3808711784841d0b5370f082974f032606","_number":1006088,"virtual_id_number":1006088,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 17:04:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2026-09-21 16:05:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2026-09-21 16:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-21 09:47:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":2,"date":"2026-09-21 16:56:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},{"value":1,"date":"2026-09-21 16:56:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-17 17:22:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-17 18:21:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2026-09-17 18:21:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-09-21 16:05:05.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-09-21 16:27:55.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-09-21 16:56:10.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"a4eb1ee02f522c900d6b601f5f91453e9e06158e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-17 17:18:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea679598b5ba5c6326dfe2b3abc1c62de9d94b97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 17:22:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba8329b5caf4f8e81a77933aa5f4467\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f54fc65fc946658dd3ffae24852d58 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"03d14574a6aacd9fd88feb5b3d14851869444d4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-17 18:17:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"61f79c3458c4bcf47dac6287553c290b528875ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-17 18:18:52.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"55d93c1b9dd259cc451fa2e45f09eebb8b5d0600","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-17 18:20:03.000000000","message":"Topic ossa-2026-039 removed","accounts_in_message":[],"_revision_number":3},{"id":"7cbbb61c2ed6514c10102ee2b6eeaa20e93ee93b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-17 18:20:07.000000000","message":"Topic set to bug/2162719","accounts_in_message":[],"_revision_number":3},{"id":"5c9d31aa6b6990c5c8f6b147534aaba7418f3bae","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-17 18:20:43.000000000","message":"Hashtag added: OSSA-2026-039","accounts_in_message":[],"_revision_number":3},{"id":"46b677ac250ed297281f8804dbb87acccf9fc15e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-17 18:22:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/973ae16f194d4a8fbacc9ba26c034655\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef240efa8134becaa97e6a20cd2db62 : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":3},{"id":"f740b5e08a45d1efe136274439e8f7eb87d04463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-18 21:27:26.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6c1cc26414301122ee0222b1ca6f42743b62dc9f","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-18 21:28:19.000000000","message":"Hashtag added: OSSA-2026-040","accounts_in_message":[],"_revision_number":4},{"id":"77afaa79b31c2a26af2acbfaddfc3d2462985ba2","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-18 21:28:45.000000000","message":"Hashtag added: CVE-2026-93852","accounts_in_message":[],"_revision_number":4},{"id":"b7b0f4f40a161097436f3bcbd6ebe62f1511ee64","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-18 21:28:51.000000000","message":"Hashtag added: CVE-2026-93854","accounts_in_message":[],"_revision_number":4},{"id":"13da1c7d0bc8421e6826906cdbf2675fbec44d63","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-18 21:28:59.000000000","message":"Hashtag removed: OSSA-2026-039","accounts_in_message":[],"_revision_number":4},{"id":"d78366d08679ce7cac6ee56c92c66457b89614f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 21:38:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dbc4758bcc6453788df403591c1452f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/867b3c85ddd3475abf6c461e77d7d207 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":4},{"id":"0770b5083942425cea8330bbd48b4365bb306eac","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2026-09-21 09:47:34.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"12d66a12057cafd3a375ab5a73c78632d216ecc7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-21 16:05:05.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"72f534f96ea687d665de5ecf6cb6b10ad31b0d0d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-09-21 16:27:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4b8eb429979e51924cfb41ff08536a0fef8e1c68","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-09-21 16:56:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"63a0c10ef96d9b540991216ee5ae5cb011a9de82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 16:56:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d41888768c02f0befd5201725b59c3d0c18e7a18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:04:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d940b7ccce9f41c1bef1793c08ce4374\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd1c38bf88b4096a57434d7b301bf25 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":4},{"id":"9db0cce16736945d76baaa90cdde3e464fdf94c6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:04:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"2c2e9f3808711784841d0b5370f082974f032606","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 17:06:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2acea2daa664f77ba5bd12a003e0553\n\n- promote-security https://zuul.opendev.org/t/openstack/build/93d4d02518774136a29d8dc1f3ca97a1 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"992854440db39687a4bb484f67437b539ff15e33","revisions":{"ea42ca1154e77c48a55d4c3e881fa10a9461c21d":{"kind":"REWORK","_number":1,"created":"2026-09-17 17:18:39.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/1006088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ossa","ref":"refs/changes/88/1006088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ossa refs/changes/88/1006088/1"}}},"commit":{"parents":[{"commit":"daf62a3b7c0224fa32861ee0805be4cfc707fb1b","subject":"Merge \"Add OSSA-2026-038\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/daf62a3b7c0224fa32861ee0805be4cfc707fb1b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-17 17:06:25.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-17 17:18:11.000000000","tz":-420},"subject":"Add OSSA-2026-039 (CVE-2026-pending)","message":"Add OSSA-2026-039 (CVE-2026-pending)\n\nChange-Id: Ic0667090d0a7e0a117aea09bbf42b54474889162\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/ea42ca1154e77c48a55d4c3e881fa10a9461c21d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/ea42ca1154e77c48a55d4c3e881fa10a9461c21d"}]},"branch":"refs/heads/master"},"691920895af817598e9b254c8da5a2f570a91af8":{"kind":"REWORK","_number":2,"created":"2026-09-17 18:17:46.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/1006088/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ossa","ref":"refs/changes/88/1006088/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ossa refs/changes/88/1006088/2"}}},"commit":{"parents":[{"commit":"daf62a3b7c0224fa32861ee0805be4cfc707fb1b","subject":"Merge \"Add OSSA-2026-038\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/daf62a3b7c0224fa32861ee0805be4cfc707fb1b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-17 17:06:25.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-17 17:21:19.000000000","tz":-420},"subject":"Add OSSA-2026-039 (CVE-2026-pending)","message":"Add OSSA-2026-039 (CVE-2026-pending)\n\nChange-Id: Ic0667090d0a7e0a117aea09bbf42b54474889162\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/691920895af817598e9b254c8da5a2f570a91af8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/691920895af817598e9b254c8da5a2f570a91af8"}]},"branch":"refs/heads/master"},"2bed56c65f945af4b475cb84e0f81e4fba1eb1e2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-17 18:18:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/1006088/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ossa","ref":"refs/changes/88/1006088/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ossa refs/changes/88/1006088/3"}}},"commit":{"parents":[{"commit":"daf62a3b7c0224fa32861ee0805be4cfc707fb1b","subject":"Merge \"Add OSSA-2026-038\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/daf62a3b7c0224fa32861ee0805be4cfc707fb1b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-17 17:06:25.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-17 18:18:52.000000000","tz":0},"subject":"Add OSSA-2026-039 (CVE-2026-pending)","message":"Add OSSA-2026-039 (CVE-2026-pending)\n\nCloses-Bug: #2162719\nChange-Id: Ic0667090d0a7e0a117aea09bbf42b54474889162\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/2bed56c65f945af4b475cb84e0f81e4fba1eb1e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/2bed56c65f945af4b475cb84e0f81e4fba1eb1e2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"992854440db39687a4bb484f67437b539ff15e33":{"kind":"REWORK","_number":4,"created":"2026-09-18 21:27:26.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/1006088/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ossa","ref":"refs/changes/88/1006088/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ossa refs/changes/88/1006088/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ossa refs/changes/88/1006088/4"}}},"commit":{"parents":[{"commit":"daf62a3b7c0224fa32861ee0805be4cfc707fb1b","subject":"Merge \"Add OSSA-2026-038\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/daf62a3b7c0224fa32861ee0805be4cfc707fb1b"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-18 21:24:24.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-18 21:24:24.000000000","tz":-420},"subject":"Add OSSA-2026-040 (CVE-2026-93852, CVE-2026-93854)","message":"Add OSSA-2026-040 (CVE-2026-93852, CVE-2026-93854)\n\nCloses-Bug: #2162719\nChange-Id: Ic0667090d0a7e0a117aea09bbf42b54474889162\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/992854440db39687a4bb484f67437b539ff15e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/992854440db39687a4bb484f67437b539ff15e33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
