)]}'
{"id":"openstack%2Fossa~1006643","triplet_id":"openstack%2Fossa~master~If3db3873ddd7a9994fabda92cc02fe3270b5087f","project":"openstack/ossa","branch":"master","topic":"ossa-2026-039","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-21 22:24:35.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2026-09-21 22:24:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If3db3873ddd7a9994fabda92cc02fe3270b5087f","subject":"OSSA-2026-039: Errata 1","status":"MERGED","created":"2026-09-21 21:12:17.000000000","updated":"2026-09-21 22:25:45.000000000","submitted":"2026-09-21 22:24:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1006643-ossa-2026-039","meta_rev_id":"63fb58d2a29ccbfee95b6de63ff3583226dfd0ae","_number":1006643,"virtual_id_number":1006643,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 22:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-21 22:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-21 22:09:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-21 21:20:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 21:46:25.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2026-09-21 21:53:11.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2026-09-21 22:09:32.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"02b60b913792a66519122e22b0f07c2a2237372d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-21 21:12:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03b958dc4a39bdd23b5349f1bbac35498066c852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 21:20:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc5b37245109440b8b7b55326ecfd8a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce24d4aa3684f05953eef469a1b098d : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a640660cb4cf3fe7f6779a10f34b672a31bbcf78","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-09-21 21:46:25.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a96af4630c505d7b29ce748b5bb55bf522fbb596","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-09-21 21:53:11.000000000","message":"Patch Set 1:\n\nDid Rolix also report one or both of the defects that got CVE assignments?","accounts_in_message":[],"_revision_number":1},{"id":"a508bc8ab377f1aad292b4072e50b5bbb80279a9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-21 21:57:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"919a87464ce4de72aa8402bc6835415b025bf4a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-21 21:58:00.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4bb79099ac8988172e47c4085a4ce5b28fe1fdcd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-21 21:58:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ccf052520378d67711ab3feb1675df40eac0c49c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-21 22:05:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"662acdddc2f68da8846c029af88ef9b48eff97df","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-09-21 22:05:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"34c18245cf7f9efffec7c762444f9e5d6690db6a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-09-21 22:09:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a7e8e5fe1866530b1cfbabe25c84eed4fbbd8189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:15:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58b63fb345444d43a51b41bda08740e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4347840be05b45af97ba3e4434286f8d : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"47703f1ffdfae86cbc180dd21b4061aa43407457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:15:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d62489ff7af6b660ea8fc18715ed5107e1240a0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:24:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db698711349f40ffbc983e93710d9718\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5209a327c4f43b997464030d3866a62 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"65ff6fc7f0d318042c764240ca7125d59fe540dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:24:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"63fb58d2a29ccbfee95b6de63ff3583226dfd0ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 22:25:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/473e9928a2d74dd181f6d31f9427d503\n\n- promote-security https://zuul.opendev.org/t/openstack/build/bfc6f171bdc744ee89e19879c298c9ac : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2f06884b2f21ba43147d0fc17f83560a70dc5c70","revisions":{"ada084705072e7448e564b74a6d0d51a9fcd7a3c":{"kind":"REWORK","_number":1,"created":"2026-09-21 21:12:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/43/1006643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ossa","ref":"refs/changes/43/1006643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ossa refs/changes/43/1006643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ossa refs/changes/43/1006643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ossa refs/changes/43/1006643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ossa refs/changes/43/1006643/1"}}},"commit":{"parents":[{"commit":"4a65f6376690abf6d0ee4975ed2bf31f693f71cf","subject":"Merge \"Add OSSA-2026-040 (CVE-2026-93852, CVE-2026-93854)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/4a65f6376690abf6d0ee4975ed2bf31f693f71cf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-21 20:31:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-21 20:34:36.000000000","tz":-420},"subject":"OSSA-2026-039: Errata 1","message":"OSSA-2026-039: Errata 1\n\nAssign CVE-2026-94571 and CVE-2026-94572.\n\nChange-Id: If3db3873ddd7a9994fabda92cc02fe3270b5087f\nRelated-Bug: #2162101\nRelated-Bug: #2162103\nRelated-Bug: #2167565\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/ada084705072e7448e564b74a6d0d51a9fcd7a3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/ada084705072e7448e564b74a6d0d51a9fcd7a3c"}]},"branch":"refs/heads/master"},"4e3499551965cd6780f705e2ba8c1abfcbc369a4":{"kind":"REWORK","_number":2,"created":"2026-09-21 21:58:00.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/43/1006643/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ossa","ref":"refs/changes/43/1006643/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ossa refs/changes/43/1006643/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ossa refs/changes/43/1006643/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ossa refs/changes/43/1006643/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ossa refs/changes/43/1006643/2"}}},"commit":{"parents":[{"commit":"4a65f6376690abf6d0ee4975ed2bf31f693f71cf","subject":"Merge \"Add OSSA-2026-040 (CVE-2026-93852, CVE-2026-93854)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/4a65f6376690abf6d0ee4975ed2bf31f693f71cf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-21 20:31:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-21 21:58:00.000000000","tz":0},"subject":"OSSA-2026-039: Errata 1","message":"OSSA-2026-039: Errata 1\n\nAssign CVE-2026-94571 and CVE-2026-94572.\n\nChange-Id: If3db3873ddd7a9994fabda92cc02fe3270b5087f\nRelated-Bug: #2162101\nRelated-Bug: #2162103\nRelated-Bug: #2167565\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/4e3499551965cd6780f705e2ba8c1abfcbc369a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/4e3499551965cd6780f705e2ba8c1abfcbc369a4"}]},"branch":"refs/heads/master"},"2f06884b2f21ba43147d0fc17f83560a70dc5c70":{"kind":"REWORK","_number":3,"created":"2026-09-21 22:05:12.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/43/1006643/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ossa","ref":"refs/changes/43/1006643/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ossa refs/changes/43/1006643/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ossa refs/changes/43/1006643/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ossa refs/changes/43/1006643/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ossa refs/changes/43/1006643/3"}}},"commit":{"parents":[{"commit":"4a65f6376690abf6d0ee4975ed2bf31f693f71cf","subject":"Merge \"Add OSSA-2026-040 (CVE-2026-93852, CVE-2026-93854)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/4a65f6376690abf6d0ee4975ed2bf31f693f71cf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-21 20:31:47.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2026-09-21 22:05:12.000000000","tz":0},"subject":"OSSA-2026-039: Errata 1","message":"OSSA-2026-039: Errata 1\n\nAssign CVE-2026-94571 and CVE-2026-94572.\n\nChange-Id: If3db3873ddd7a9994fabda92cc02fe3270b5087f\nRelated-Bug: #2162101\nRelated-Bug: #2162103\nRelated-Bug: #2167565\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/2f06884b2f21ba43147d0fc17f83560a70dc5c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ossa/commit/2f06884b2f21ba43147d0fc17f83560a70dc5c70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
