)]}'
{"id":"openstack%2Fovn-bgp-agent~915313","triplet_id":"openstack%2Fovn-bgp-agent~master~I176c78cc3409ca72d2f0819078593a4c4ac768ec","project":"openstack/ovn-bgp-agent","branch":"master","attention_set":{"34799":{"account":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"last_update":"2025-02-24 18:51:47.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2025-02-24 18:51:47.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"I176c78cc3409ca72d2f0819078593a4c4ac768ec","subject":"Ignore Bridges where Interface is down","status":"NEW","created":"2024-04-09 09:31:49.000000000","updated":"2025-02-24 18:51:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b03af186bb9c36a2ed8a81a42c18787d159480c9","_number":915313,"virtual_id_number":915313,"owner":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-06 10:23:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-09 11:10:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-04 11:01:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2024-08-15 15:13:37.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"bfffd1aed45c9351d6569c227b6e92053cc34f83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"date":"2024-04-09 09:31:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"677fe35888b881a7cd1dbd41e792cfbd04b35614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 11:10:17.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b25765d4a18b4650affc6ad32bd937c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd0c9650169146a8981d58be9b9fdc38 : FAILURE in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7089f47174684b5da3572c0a59b13a43 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e1d769f0e6e4ceb8683b7d179f1a558 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b17975181584b2b89c4ef6e5a2c08eb : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0236b3164d2043ba9ab4d7c2619a4b83 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe46b5298c845e2a8999de7db470341 : SUCCESS in 2m 48s\n- openstack-tox-functional-with-sudo https://zuul.opendev.org/t/openstack/build/7698cdcef0d142abb2e83ea312acc56b : SUCCESS in 2m 47s\n- ovn-bgp-agent-tempest-plugin https://zuul.opendev.org/t/openstack/build/6c5b14917bee4a79b622f887cc426831 : SUCCESS in 1h 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"978551a2c15c4b6688cf7db0634ac22e542f36e0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-06-04 11:01:58.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"50c812574caa234de7c7f775d58985f8c012592f","tag":"autogenerated:gerrit:restore","author":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"date":"2024-08-14 12:13:55.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"f4434cde3f31b68372d79e5b6d8000beae8c2671","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"date":"2024-08-14 12:16:29.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"16b2ddb5f565436ad98b159908dd1bc7c887a80b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 14:08:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb1bdc592dae4c129b301b461e01701e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4fd3927f21473c8c25b26f373acddb : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de68efb40ad64a89a87b8e2b3c085aad : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf36746bc6ac40639dfc1bcf401326ea : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/563a9b5cb1db499283250acd6a40c2a0 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b9f1ba6fccb4d6eb482b44d8766c0cc : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe34b9d4a6aa49419195c41b16b0392e : SUCCESS in 2m 54s\n- openstack-tox-functional-with-sudo https://zuul.opendev.org/t/openstack/build/0b1410c5183c4884b64b9395e355bbb5 : SUCCESS in 2m 53s\n- ovn-bgp-agent-tempest-plugin https://zuul.opendev.org/t/openstack/build/468dd0b64dbf45b99ace464601f96c35 : FAILURE in 1h 51m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60af0842fc07948a4b98bcf6c068eaca03bc26c5","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2024-08-15 15:13:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9331f98684b894467a4cb8b36efd7f85057d1fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"date":"2025-02-06 07:42:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"60abe7e62205ce2ded29c733fac399394480976d","author":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"date":"2025-02-06 07:51:54.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"43ebf0911c2a7cebabcb51cb0127ff95c5f2c6a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"date":"2025-02-06 07:53:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"13917b8f4aa50e77ee0469f52cd89417dc3a19ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 10:23:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2db8bf409040a7beeab9eca980bc1b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f15fc2cb695c4afcb7e7deb7772fc87b : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8ae4cca6b3a4268ba5de320e8bc5f8c : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/245862e7d3b84c4a84aaa7785ffd9493 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd6c798060984774a298cc82314f3c1b : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/027dae0e2ba94fe18647e45f81b78f17 : SUCCESS in 2m 41s\n- openstack-tox-functional-with-sudo https://zuul.opendev.org/t/openstack/build/d3961bf1754e47cfb70fb3a81c38212b : SUCCESS in 7m 07s\n- ovn-bgp-agent-tempest-plugin-multinode https://zuul.opendev.org/t/openstack/build/1cb0be16f2b042d7a453ad1a4148b6d9 : SUCCESS in 2h 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b03af186bb9c36a2ed8a81a42c18787d159480c9","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2025-02-24 18:51:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"43a55bea1c1e4b6e5c50cbc03cec76db0f9d84ed","revisions":{"f7d797135d96a793b83aae6a28da4257f7f8ff61":{"kind":"REWORK","_number":1,"created":"2024-04-09 09:31:49.000000000","uploader":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"ref":"refs/changes/13/915313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovn-bgp-agent","ref":"refs/changes/13/915313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/1"}}},"commit":{"parents":[{"commit":"bdbf573ae41933e2ec3985c2a3daf413177ec666","subject":"Merge \"Fix placement of lsp when external_ids not in sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/bdbf573ae41933e2ec3985c2a3daf413177ec666"}]}],"author":{"name":"Justin Lamp","email":"justin.lamp@netways.de","date":"2024-04-09 09:30:33.000000000","tz":120},"committer":{"name":"Justin Lamp","email":"justin.lamp@netways.de","date":"2024-04-09 09:30:33.000000000","tz":120},"subject":"Ignore Bridges where Interface is down","message":"Ignore Bridges where Interface is down\n\nWhen the interface of a bridge found in the bridge-mapping is down and not up, issue a warning and continue the sync without it.\n\nRelated-bug: #2060453\nChange-Id: I176c78cc3409ca72d2f0819078593a4c4ac768ec\nSigned-off-by: Justin Lamp \u003cjustin.lamp@netways.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/f7d797135d96a793b83aae6a28da4257f7f8ff61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/f7d797135d96a793b83aae6a28da4257f7f8ff61"}]},"branch":"refs/heads/master"},"566e9f9d5dfc08c672e00ebf0cfe5563b7cd9654":{"kind":"REWORK","_number":2,"created":"2024-08-14 12:16:29.000000000","uploader":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"ref":"refs/changes/13/915313/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovn-bgp-agent","ref":"refs/changes/13/915313/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/2"}}},"commit":{"parents":[{"commit":"bdbf573ae41933e2ec3985c2a3daf413177ec666","subject":"Merge \"Fix placement of lsp when external_ids not in sync\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/bdbf573ae41933e2ec3985c2a3daf413177ec666"}]}],"author":{"name":"Justin Lamp","email":"justin.lamp@netways.de","date":"2024-04-09 09:30:33.000000000","tz":120},"committer":{"name":"Justin Lamp","email":"justin.lamp@netways.de","date":"2024-08-14 12:16:29.000000000","tz":0},"subject":"Ignore Bridges where Interface is down","message":"Ignore Bridges where Interface is down\n\nWhen the interface of a bridge found in the bridge-mapping is down and not up, issue a warning and continue the sync without it.\n\nRelated-bug: #2060453\nChange-Id: I176c78cc3409ca72d2f0819078593a4c4ac768ec\nSigned-off-by: Justin Lamp \u003cjustin.lamp@netways.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/566e9f9d5dfc08c672e00ebf0cfe5563b7cd9654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/566e9f9d5dfc08c672e00ebf0cfe5563b7cd9654"}]},"branch":"refs/heads/master"},"9e5be225cf207bdca3faae9f4a2e576ae50767ed":{"kind":"REWORK","_number":3,"created":"2025-02-06 07:42:48.000000000","uploader":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"ref":"refs/changes/13/915313/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovn-bgp-agent","ref":"refs/changes/13/915313/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/3"}}},"commit":{"parents":[{"commit":"d4f9b7ef1bc799bcc56939277972f1b7f21582cf","subject":"Merge \"Add option to avoid VRF removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/d4f9b7ef1bc799bcc56939277972f1b7f21582cf"}]}],"author":{"name":"Justin Lamp","email":"justin.lamp@netways.de","date":"2024-04-09 09:30:33.000000000","tz":120},"committer":{"name":"Justin Lamp","email":"justin.lamp@netways.de","date":"2025-02-06 07:41:03.000000000","tz":60},"subject":"Ignore Bridges where Interface is down","message":"Ignore Bridges where Interface is down\n\nWhen the interface of a bridge found in the bridge-mapping is down and\nnot up, issue a warning and continue the sync without it.\n\nRelated-bug: #2060453\nChange-Id: I176c78cc3409ca72d2f0819078593a4c4ac768ec\nSigned-off-by: Justin Lamp \u003cjustin.lamp@netways.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/9e5be225cf207bdca3faae9f4a2e576ae50767ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/9e5be225cf207bdca3faae9f4a2e576ae50767ed"}]},"branch":"refs/heads/master"},"43a55bea1c1e4b6e5c50cbc03cec76db0f9d84ed":{"kind":"REWORK","_number":4,"created":"2025-02-06 07:53:38.000000000","uploader":{"_account_id":34799,"name":"Justin Lamp","email":"justin.lamp96@gmail.com","username":"justinlamp","status":"Netways GmbH"},"ref":"refs/changes/13/915313/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovn-bgp-agent","ref":"refs/changes/13/915313/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovn-bgp-agent refs/changes/13/915313/4"}}},"commit":{"parents":[{"commit":"d4f9b7ef1bc799bcc56939277972f1b7f21582cf","subject":"Merge \"Add option to avoid VRF removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/d4f9b7ef1bc799bcc56939277972f1b7f21582cf"}]}],"author":{"name":"Justin Lamp","email":"justin.lamp@netways.de","date":"2024-04-09 09:30:33.000000000","tz":120},"committer":{"name":"Justin Lamp","email":"justin.lamp@netways.de","date":"2025-02-06 07:53:30.000000000","tz":60},"subject":"Ignore Bridges where Interface is down","message":"Ignore Bridges where Interface is down\n\nWhen the interface of a bridge found in the bridge-mapping is down and\nnot up, issue a warning and continue the sync without it.\n\nRelated-bug: #2060453\nChange-Id: I176c78cc3409ca72d2f0819078593a4c4ac768ec\nSigned-off-by: Justin Lamp \u003cjustin.lamp@netways.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/43a55bea1c1e4b6e5c50cbc03cec76db0f9d84ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovn-bgp-agent/commit/43a55bea1c1e4b6e5c50cbc03cec76db0f9d84ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
