)]}'
{"id":"openstack%2Fovsdbapp~633670","triplet_id":"openstack%2Fovsdbapp~master~I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef","project":"openstack/ovsdbapp","branch":"master","topic":"update-hacking","hashtags":[],"change_id":"I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef","subject":"Update hacking for Python3","status":"MERGED","created":"2019-01-29 08:19:24.000000000","updated":"2020-04-02 21:55:30.000000000","submitted":"2020-04-02 21:51:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"633670-1585864266147-9ed293ab","meta_rev_id":"a9fa7150c3a50ab59b927e1400d1656725771957","_number":633670,"virtual_id_number":633670,"owner":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-04-02 13:26:54.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 21:51:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2020-04-02 13:28:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-04-02 09:51:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-04-02 19:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2020-04-02 07:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-04-02 19:11:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-28 16:35:56.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-03-31 02:37:20.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-03-31 13:21:59.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-02 07:15:57.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-04-02 09:51:10.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-04-02 13:26:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-02 13:28:07.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-04-02 19:11:33.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-02 21:51:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39383dd38b08de87c3acf4e3d97bf008aece6956","author":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"date":"2019-01-29 08:19:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"689278aa288056b9571fc45d54135f68d3f76e53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-29 08:30:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/70/633670/1/check/openstack-tox-pep8/1aa602f/ : FAILURE in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/70/633670/1/check/openstack-tox-py27/f9634b7/ : SUCCESS in 3m 16s\n- openstack-tox-py35 http://logs.openstack.org/70/633670/1/check/openstack-tox-py35/07cfc73/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/70/633670/1/check/openstack-tox-py36/14e0977/ : SUCCESS in 3m 19s\n- requirements-check http://logs.openstack.org/70/633670/1/check/requirements-check/049ec6f/ : SUCCESS in 2m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/633670/1/check/openstack-tox-lower-constraints/85121da/ : SUCCESS in 4m 43s\n- openstack-tox-functional http://logs.openstack.org/70/633670/1/check/openstack-tox-functional/c4384f3/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"70d31c47552f63c5a543bcb0c2fc847c06640ccf","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-03-20 14:54:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3f0850a8cd9eeb5421ad0df1bfd9eae06afa230","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-20 17:13:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/633670/2/check/openstack-tox-pep8/632e171/ : SUCCESS in 4m 14s\n- openstack-tox-py27 http://logs.openstack.org/70/633670/2/check/openstack-tox-py27/4d490bc/ : SUCCESS in 4m 02s\n- openstack-tox-py35 http://logs.openstack.org/70/633670/2/check/openstack-tox-py35/c907fcf/ : SUCCESS in 4m 00s\n- openstack-tox-py36 http://logs.openstack.org/70/633670/2/check/openstack-tox-py36/70feda6/ : SUCCESS in 3m 41s\n- requirements-check http://logs.openstack.org/70/633670/2/check/requirements-check/236b720/ : SUCCESS in 2m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/633670/2/check/openstack-tox-lower-constraints/57bb0a1/ : SUCCESS in 4m 43s\n- openstack-tox-functional http://logs.openstack.org/70/633670/2/check/openstack-tox-functional/3a96b4d/ : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 http://logs.openstack.org/70/633670/2/check/openstack-tox-functional-py36/ebda48b/ : SUCCESS in 23m 50s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/70/633670/2/check/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/6348ee3/ : SUCCESS in 2h 14m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1b51bff9804210c4861aa2c583129e6e2c671c6b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-03-21 11:54:35.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"502964f2cfc75344ee54148524e99f63a3e0aa88","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-03-28 16:35:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e06110a0f58d941705e08c59f5d93be1e52f8e56","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-19 08:53:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"213d45a92636de611b80bf972741e95ee9937ee0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-27 14:54:11.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nIt\u0027s #friday, i\u0027ll do a quick update.\n\nI also noticed ovsdbapp still supports py2, should probably add a low-priority bug to make it py3 only following the same steps we did in other repos, as eventually py2 specific things could stop working.","accounts_in_message":[],"_revision_number":2},{"id":"1df1cb493220dac3435760ef32613785ead733cf","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-27 14:54:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"65990893e269deb74125fa321a6f29e7f2292a2b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-03-27 15:03:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4135aee6e7d80bf731d62da3cd9824db7b43c18f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-03-27 15:04:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c1469a255964bacfce8c493192fea923a73d0a92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 15:36:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1429febaf8cd4cc98056c26e8895c9b0 : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/098534f1dcf441b7ab8c009f2ba1234b : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4137714ab01340c4a3b585524968abb1 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f874760b8b2f41ad8ea751aa86d9c06d : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43652d2bec3c486e862126dd74c1859c : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/307703dd1d71400cb0fff7f97a7fffe5 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/affeeea31e18462cad8080953588a378 : SUCCESS in 5m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d95b1366eee6428ea2fef822648084bb : SUCCESS in 8m 58s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/e41981312b8e4cf79cbd6205d8edcc42 : FAILURE in 35m 18s","accounts_in_message":[],"_revision_number":3},{"id":"f55bb30ed5391b722242f846005b49dea1ab2763","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-03-27 16:10:11.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d3edeb87574904f3c999b3fec42d7b2c3107aeed","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-03-27 16:10:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"df45888ca548a50808cf80e8ce798745300df0ea","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-03-27 16:20:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ca10051b63d269deebcb14a08349a0ff30e25c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 16:49:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7a3b6b6eb47b4f7c961674324899b271 : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25f3ae19c1fb4c14a31d8b5c3f2ad018 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b27a2aeebba4d60ad129433e4ba6b57 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6429b207547b4144892ed399700cefe6 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d61a188adf14d2791c98431f4df79fa : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb341cb90b184780a27d7dc9f65b501c : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2b5cefe31f1422f9f3ba6f6e914daec : SUCCESS in 8m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a051742060f24c6bac030b292bbeba9a : SUCCESS in 7m 12s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/7373d606835e4b099976ea28b9e5de8e : FAILURE in 32m 31s","accounts_in_message":[],"_revision_number":4},{"id":"775c3430ff331c62ecc452dfbd0d40d36ce08c69","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-27 16:52:54.000000000","message":"Patch Set 4:\n\nFailing in way I\u0027ve seen in other repo recently:\n\npylint 2.3.0 has requirement astroid\u003e\u003d2.2.0, but you have astroid 2.1.0.\n\nSomething got bumped.\n\nAnd regarding the lambda, could always just ignore E731 as neutron does, although I personally like functions over lambda :)","accounts_in_message":[],"_revision_number":4},{"id":"fa3aaa8910c1579d7315ef77a33520957596c7b0","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-03-27 17:57:39.000000000","message":"Patch Set 4:\n\ninteresting since test-requirements has pylint\u003d\u003d1.9.2...","accounts_in_message":[],"_revision_number":4},{"id":"c6b0f1a4c7003101bc275fbc6385def582621d71","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-03-27 18:19:56.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"3a389ab95f3a50b5aebd620968a8c85718934cdc","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-03-27 18:20:19.000000000","message":"Patch Set 5:\n\ntrying out the pylint/astroid versions from neutron repo","accounts_in_message":[],"_revision_number":5},{"id":"17fdc00db0938ecec2166da64bd53c9d051cc202","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 19:00:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/00df992c62cc4dc8b80562a2f7946c11 : SUCCESS in 4m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43b869ece0a9467cab304f6f02e9f392 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7478a51be0ef4b8593b279e4d1086d07 : FAILURE in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad2db6875fdc46b3ae6adfa4accbed06 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73239ae8828c433ca3b79beefd3eb617 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5749f25094a44c9f96028633282a98ed : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec49d4e312d47b5b9768908e6ffeee0 : SUCCESS in 6m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cddee8b4f3a749daafd99026a34a37a0 : SUCCESS in 7m 39s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/0ec45b1849aa464eabd1ec1d248ded86 : FAILURE in 34m 22s","accounts_in_message":[],"_revision_number":5},{"id":"b7757c3bb9059080fa48f4cd3bb7ecae28bb7349","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-03-27 19:56:00.000000000","message":"Patch Set 5:\n\nOk, so just realizing that I was silly since the test that was running was a networking-ovn test so not using my test-requirements and I broke the tests that were actually in ovsdbapp...","accounts_in_message":[],"_revision_number":5},{"id":"59f6b5df8c85255c1d878e5555f282c9accb34fb","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-03-27 19:58:53.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"0d18794169929669e0e8d272625e0a4f7f88460d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-27 20:31:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d899a9f5b97e4547b9fcdc0cd1d5cac2 : SUCCESS in 2m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/232c37a4a29542568e96ab55a7f6aa05 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3428871e5d34498b3ab0ab8489c99f8 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/173b8c530f114982b2a85278c5eafb8b : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e989c9f85178448cb9e7e25a60ade39f : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9273821649742aa9f2dacf6504f2647 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53c219e626941ef87194992abc160ad : SUCCESS in 5m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e09420856b04449ebcbcf9bd8f744730 : SUCCESS in 5m 50s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/6a2f251a3f6647508e08467f4a28b418 : FAILURE in 31m 34s","accounts_in_message":[],"_revision_number":6},{"id":"f004e72e6bf3262bf214db2c39cc6ddf369e959c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-30 21:57:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"00f06bedbf78922c88ca240a5a36c7a036fc7d6e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-30 21:58:11.000000000","message":"Topic changed from 633670 to update-hacking","accounts_in_message":[],"_revision_number":7},{"id":"782cab2a1cae18f2016dd0c1666e5c2145b2afb3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-30 23:09:52.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e62f9d66fcae2fa063641d50f24af83aa2e72086","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-30 23:18:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"26b4b4f5c71e10507fb18698d2e1569a19b34fab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 01:45:02.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/da40c35e41b34930877719f8da167956 : SUCCESS in 2m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70e849e1f73e48dd8e2a245a6eeb6508 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1097fd776ad4780b7bc79c276d2c8e3 : FAILURE in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a3ea43d18154fd5bb192bffd60f5765 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27b4de7f8f3a4675991d6c77d4ef9318 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49c86f0c1a124559b12a2c333cfed105 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5bd1c50479149a0a6ff761f133c7091 : SUCCESS in 5m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f064b2997d3c4a51a4eae7450db9fc62 : SUCCESS in 6m 01s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/62113661817f4c189507f5d9060962dc : SUCCESS in 2h 32m 56s","accounts_in_message":[],"_revision_number":8},{"id":"c62988a40cc4d7b695f36a1d52dae580ba84a53b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-31 02:37:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dd344f5d2b239f6cf120bd85c47a63dfadb58df6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-31 13:21:59.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nA couple of pep8 errors should be addressed:\n\n2020-03-30 23:14:06.649679 | ubuntu-bionic |   import imp\n2020-03-30 23:14:06.649717 | ubuntu-bionic | \u001b[7;33m************* Module ovsdbapp.exceptions\u001b[0m\n2020-03-30 23:14:06.649748 | ubuntu-bionic | ovsdbapp/exceptions.py:32:12: R1720: \u001b[1;3;35mUnnecessary \"else\" after \"raise\"\u001b[0m (\u001b[1;3;35mno-else-raise\u001b[0m)\n2020-03-30 23:14:06.649783 | ubuntu-bionic | \u001b[7;33m************* Module ovsdbapp.venv\u001b[0m\n2020-03-30 23:14:06.649815 | ubuntu-bionic | ovsdbapp/venv.py:224:8: W0125: \u001b[35mUsing a conditional statement with a constant value\u001b[0m (\u001b[35musing-constant-test\u001b[0m)\n2020-03-30 23:14:06.649843 | ubuntu-bionic | \u001b[7;33m************* Module ovsdbapp.backend.ovs_idl.rowview\u001b[0m\n2020-03-30 23:14:06.649881 | ubuntu-bionic | ovsdbapp/backend/ovs_idl/rowview.py:25:15: E0712: \u001b[1;31mCatching an exception which doesn\u0027t inherit from Exception: NotImplemented\u001b[0m (\u001b[1;31mcatching-non-exception\u001b[0m)","accounts_in_message":[],"_revision_number":8},{"id":"6dc2e6139f6550ce2673bcf4c66d22cdaeef4eb4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-31 23:52:49.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b860d094be4095a7059440fb4fd761ff1ebce5a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 02:29:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/47f8a18094c144dbbbf5ca47589c5a20 : SUCCESS in 2m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c7e87ab73844046afe3c0c9b0a7762d : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c26c11f6bf724dac8c6ea0ca9b98cce8 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b203d58d43e4faba07a211625e11718 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0ce865edb354d71ab88d0e9bf3c9270 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21bb0e68d0f14a9eb94d135613a86d50 : FAILURE in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2684714297994c5db725f4a4b3171b28 : SUCCESS in 4m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e3373a22de544ac790f5d09f65afb39b : SUCCESS in 5m 56s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/343d26b35a434c6da1258dbe5ca2bae4 : SUCCESS in 2h 35m 13s","accounts_in_message":[],"_revision_number":9},{"id":"30e6619759f81b26590335a7625197887d8e8784","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-04-01 08:07:45.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)\n\nLooks good :)","accounts_in_message":[],"_revision_number":9},{"id":"b5efd00a6b073beffd18e42301510389421c0fd5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-01 08:41:55.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"fc42bcdf27a591f842ae6e50acd2bc16dbf48e32","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-01 14:07:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7a52c1d4a07986dbf403c787f599737f9f3a504d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-01 14:07:25.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"63d77076d53cf16f8b7e7981a8db6d4008ace5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 17:03:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f75d57cc36f4cdf9fb86da2f141f94a : SUCCESS in 3m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3165012efe540c19258a74f774c99e7 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a9f470320a491eaaced3a8b7453572 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63bb888a78204c6f8a29b325c4ac8043 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f712a4b0e6024dadbcdf7232bd0c79a0 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d34151d378141c996948b37c6aee7d2 : FAILURE in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f7e4c2ff5f14aafb812d0f4a18008fb : SUCCESS in 4m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ccc6961e71bc4cd3a2fb1f3f8cc5664e : SUCCESS in 6m 32s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/4580f56f480c4f35a5b4bf40db1ead50 : SUCCESS in 2h 50m 49s","accounts_in_message":[],"_revision_number":10},{"id":"7e486d617ee56f9def35a1d596ec5c9c0ab55223","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-04-02 07:15:57.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"02bd0cf0d013708bf5cc3fd932c0d7353e4aa585","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2020-04-02 09:51:10.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"38fb5efbc02791e58353a70020212c7619935d9e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-02 10:01:53.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"95a17f96f2dfff59891de286975e1da55600a9c2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-02 13:26:54.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a8eed7a098a73284274d6da80c04cba7ec32e4d1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-02 13:28:07.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e59cc8a13cd2eab1ead83d9f10609292978e8404","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-02 19:11:33.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"c74602470e79f2a1b3584970c1f0388ecefcff39","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-02 19:11:48.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"5ecbd3fc68dc5da860daccbcc684ad0da5d0b2c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 19:22:15.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"0048db57a3580499230f612857a5290c36630f44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:51:05.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/de3a4e4057aa405f94bd9bc2474612ed : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a33d88aa3284b019f23591d853ebdba : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9baf97b664394733aea6be64ec69737c : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f02c3f7bf1ff4cbc84b08d0c7d19f8a5 : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9b66e051004c4af390c80524732c0428 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4636de06f940518fbeb421b66564b1 : SUCCESS in 7m 11s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8fdfd8f883924b6a93cb808a1e841f53 : SUCCESS in 8m 10s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/969807387121472d9b2343f1799615fe : SUCCESS in 2h 19m 41s","accounts_in_message":[],"_revision_number":10},{"id":"191f73b8318cdbae3b07bca0e816740e949bbaad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:51:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"a9fa7150c3a50ab59b927e1400d1656725771957","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:55:30.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b1a19127ee4e5a89d52d7aaf605d08 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e6c2f1fed11dfff56b0f8308bfa00b9f8429475b","revisions":{"c9e068437760b5c3bb452d5d1572826693de5cdb":{"kind":"REWORK","_number":1,"created":"2019-01-29 08:19:24.000000000","uploader":{"_account_id":29423,"name":"zhouxinyong","email":"zhouxinyong@inspur.com","username":"zhouxinyong"},"ref":"refs/changes/70/633670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/633670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/1"}}},"commit":{"parents":[{"commit":"039da03fe4005161e9af2131a43f6355eb1cf1d0","subject":"Merge \"Allow read-only Commands to bypass txns in execute()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/039da03fe4005161e9af2131a43f6355eb1cf1d0"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:17:28.000000000","tz":480},"committer":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:19:04.000000000","tz":480},"subject":"Fix version in test-requirements.txt","message":"Fix version in test-requirements.txt\n\nChange-Id: I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c9e068437760b5c3bb452d5d1572826693de5cdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c9e068437760b5c3bb452d5d1572826693de5cdb"}]},"branch":"refs/heads/master"},"063c725ee26a1b97059faf5d66092e0a476c0792":{"kind":"REWORK","_number":2,"created":"2019-03-20 14:54:07.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/70/633670/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/633670/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/2"}}},"commit":{"parents":[{"commit":"c52e53b1bc3bb2d783ec1716094893fd2dd77eb0","subject":"Add mock.patch.stopall cleanup handler to base test class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c52e53b1bc3bb2d783ec1716094893fd2dd77eb0"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:17:28.000000000","tz":480},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-03-20 14:53:42.000000000","tz":-300},"subject":"Update hacking version in test-requirements.txt","message":"Update hacking version in test-requirements.txt\n\nChange-Id: I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/063c725ee26a1b97059faf5d66092e0a476c0792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/063c725ee26a1b97059faf5d66092e0a476c0792"}]},"branch":"refs/heads/master"},"c62772e3a4a633174da6e9834c0d19c5702d8bb9":{"kind":"REWORK","_number":3,"created":"2020-03-27 14:54:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/633670/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/633670/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/3"}}},"commit":{"parents":[{"commit":"c52e53b1bc3bb2d783ec1716094893fd2dd77eb0","subject":"Add mock.patch.stopall cleanup handler to base test class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c52e53b1bc3bb2d783ec1716094893fd2dd77eb0"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:17:28.000000000","tz":480},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-27 14:54:17.000000000","tz":-240},"subject":"Update hacking version in test-requirements.txt","message":"Update hacking version in test-requirements.txt\n\nFixed related code warnings.\n\nChange-Id: I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c62772e3a4a633174da6e9834c0d19c5702d8bb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c62772e3a4a633174da6e9834c0d19c5702d8bb9"}]},"branch":"refs/heads/master"},"b22cc180524a02682405bd1bfb05bd56ccd55b01":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-27 16:10:19.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/70/633670/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/633670/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/4"}}},"commit":{"parents":[{"commit":"85d632327b0daad854151289c190dee0dcc680d5","subject":"Merge \"Do not try to execute transaction if command list is empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/85d632327b0daad854151289c190dee0dcc680d5"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:17:28.000000000","tz":480},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-03-27 16:10:19.000000000","tz":0},"subject":"Update hacking version in test-requirements.txt","message":"Update hacking version in test-requirements.txt\n\nFixed related code warnings.\n\nChange-Id: I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b22cc180524a02682405bd1bfb05bd56ccd55b01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b22cc180524a02682405bd1bfb05bd56ccd55b01"}]},"branch":"refs/heads/master"},"c8e7b4fad9078f17b14bec7317a16c43e4a6f67d":{"kind":"REWORK","_number":5,"created":"2020-03-27 18:19:56.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/70/633670/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/633670/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/5"}}},"commit":{"parents":[{"commit":"85d632327b0daad854151289c190dee0dcc680d5","subject":"Merge \"Do not try to execute transaction if command list is empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/85d632327b0daad854151289c190dee0dcc680d5"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:17:28.000000000","tz":480},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-03-27 18:18:53.000000000","tz":0},"subject":"Update hacking version in test-requirements.txt","message":"Update hacking version in test-requirements.txt\n\nFixed related code warnings.\n\nChange-Id: I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c8e7b4fad9078f17b14bec7317a16c43e4a6f67d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c8e7b4fad9078f17b14bec7317a16c43e4a6f67d"}]},"branch":"refs/heads/master"},"a8a5033c1b6af52a1b5f45220ab1bf4baa7e0d7e":{"kind":"REWORK","_number":6,"created":"2020-03-27 19:58:53.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/70/633670/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/633670/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/6"}}},"commit":{"parents":[{"commit":"85d632327b0daad854151289c190dee0dcc680d5","subject":"Merge \"Do not try to execute transaction if command list is empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/85d632327b0daad854151289c190dee0dcc680d5"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:17:28.000000000","tz":480},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-03-27 19:58:11.000000000","tz":0},"subject":"Update hacking version in test-requirements.txt","message":"Update hacking version in test-requirements.txt\n\nFixed related code warnings.\n\nChange-Id: I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a8a5033c1b6af52a1b5f45220ab1bf4baa7e0d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a8a5033c1b6af52a1b5f45220ab1bf4baa7e0d7e"}]},"branch":"refs/heads/master"},"200d77c194308f07a87b3adf52165ac4323b9136":{"kind":"REWORK","_number":7,"created":"2020-03-30 21:57:56.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/633670/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/633670/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/7"}}},"commit":{"parents":[{"commit":"85d632327b0daad854151289c190dee0dcc680d5","subject":"Merge \"Do not try to execute transaction if command list is empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/85d632327b0daad854151289c190dee0dcc680d5"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:17:28.000000000","tz":480},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 21:55:26.000000000","tz":-240},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFixed related code warnings.\n\nChange-Id: I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/200d77c194308f07a87b3adf52165ac4323b9136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/200d77c194308f07a87b3adf52165ac4323b9136"}]},"branch":"refs/heads/master"},"7e23da675d2107b8c4e5033237c0dec6caa4d63c":{"kind":"REWORK","_number":8,"created":"2020-03-30 23:09:52.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/633670/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/633670/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/8"}}},"commit":{"parents":[{"commit":"85d632327b0daad854151289c190dee0dcc680d5","subject":"Merge \"Do not try to execute transaction if command list is empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/85d632327b0daad854151289c190dee0dcc680d5"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:17:28.000000000","tz":480},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-30 23:09:22.000000000","tz":-240},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFixed related code warnings and updated .pylintrc.\n\nChange-Id: I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7e23da675d2107b8c4e5033237c0dec6caa4d63c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7e23da675d2107b8c4e5033237c0dec6caa4d63c"}]},"branch":"refs/heads/master"},"fc8768bd06ea889f6c4bb989b84ce0b691d04932":{"kind":"REWORK","_number":9,"created":"2020-03-31 23:52:49.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/633670/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/633670/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/9"}}},"commit":{"parents":[{"commit":"85d632327b0daad854151289c190dee0dcc680d5","subject":"Merge \"Do not try to execute transaction if command list is empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/85d632327b0daad854151289c190dee0dcc680d5"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:17:28.000000000","tz":480},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-03-31 23:52:39.000000000","tz":-240},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFixed related code warnings and updated .pylintrc.\n\nChange-Id: I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fc8768bd06ea889f6c4bb989b84ce0b691d04932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fc8768bd06ea889f6c4bb989b84ce0b691d04932"}]},"branch":"refs/heads/master"},"e6c2f1fed11dfff56b0f8308bfa00b9f8429475b":{"kind":"REWORK","_number":10,"created":"2020-04-01 14:07:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/70/633670/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/633670/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/633670/10"}}},"commit":{"parents":[{"commit":"85d632327b0daad854151289c190dee0dcc680d5","subject":"Merge \"Do not try to execute transaction if command list is empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/85d632327b0daad854151289c190dee0dcc680d5"}]}],"author":{"name":"zhouxinyong","email":"zhouxinyong@inspur.com","date":"2019-01-29 08:17:28.000000000","tz":480},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2020-04-01 14:06:32.000000000","tz":-240},"subject":"Update hacking for Python3","message":"Update hacking for Python3\n\nThe repo is Python 3 now, so update hacking to version 3.0 which\nsupports Python 3.\n\nFixed related code warnings and updated .pylintrc.\n\nChange-Id: I5e82d29e0d6f09b98bebdd5dbcc34b01512b28ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e6c2f1fed11dfff56b0f8308bfa00b9f8429475b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e6c2f1fed11dfff56b0f8308bfa00b9f8429475b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
