)]}'
{"id":"openstack%2Fovsdbapp~654261","triplet_id":"openstack%2Fovsdbapp~master~I90dfdb5b5fe625ca8af768dfe9dac10f6d3c1430","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1825740","hashtags":[],"change_id":"I90dfdb5b5fe625ca8af768dfe9dac10f6d3c1430","subject":"Allow match_ip in LrNatDelCommand to be IP network","status":"MERGED","created":"2019-04-21 19:01:21.000000000","updated":"2019-05-31 22:11:47.000000000","submitted":"2019-05-31 22:11:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654261-1559340707746-7714beb8","meta_rev_id":"654a11a070743c7da63fc004a961e32ff23fa0a0","_number":654261,"virtual_id_number":654261,"owner":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":2,"date":"2019-05-31 22:11:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-02 15:57:25.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},{"value":2,"date":"2019-05-31 19:44:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2019-05-31 19:52:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2019-05-31 19:52:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"value":0,"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-21 19:01:34.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-04-21 19:03:42.000000000","updated_by":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"reviewer":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"state":"REVIEWER"},{"updated":"2019-04-21 19:03:42.000000000","updated_by":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"reviewer":{"_account_id":1653,"name":"garyk","email":"gkotton@vmware.com","username":"garyk"},"state":"REVIEWER"},{"updated":"2019-04-21 19:03:42.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-04-21 19:03:42.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-04-21 19:03:42.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-04-21 19:03:42.000000000","updated_by":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"reviewer":{"_account_id":6854,"name":"YAMAMOTO Takashi","email":"yamamoto@midokura.com","username":"yamamoto"},"state":"REVIEWER"},{"updated":"2019-04-21 19:03:42.000000000","updated_by":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"reviewer":{"_account_id":8788,"name":"Miguel Angel Ajo","email":"mangelajo@redhat.com","username":"mangelajo"},"state":"REVIEWER"},{"updated":"2019-04-21 19:03:42.000000000","updated_by":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"reviewer":{"_account_id":10237,"name":"Numan Siddique","email":"nusiddiq@redhat.com","username":"numansiddique"},"state":"REVIEWER"},{"updated":"2019-04-21 19:03:42.000000000","updated_by":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"reviewer":{"_account_id":20229,"name":"Omer Anson","email":"omer.anson@toganetworks.com","username":"oanson"},"state":"REVIEWER"},{"updated":"2019-04-22 18:00:16.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-05-02 15:57:25.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-05-31 19:44:48.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-05-31 19:52:59.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-05-31 22:11:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1a5aefda1e521c56c67808ba5ad27b66612d5ff1","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-21 19:01:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8062814b8bfd631e88c6d37f675ccaf7e127dd5a","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-04-21 19:01:34.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"724e9bb4d01419bc30bec233c60460f81cfd9195","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 21:33:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/61/654261/1/check/openstack-tox-pep8/5c301e9/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/61/654261/1/check/openstack-tox-py27/af9e35b/ : SUCCESS in 2m 44s\n- openstack-tox-py35 http://logs.openstack.org/61/654261/1/check/openstack-tox-py35/0318393/ : SUCCESS in 3m 01s\n- openstack-tox-py36 http://logs.openstack.org/61/654261/1/check/openstack-tox-py36/42900b8/ : SUCCESS in 2m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/654261/1/check/openstack-tox-lower-constraints/43f2e43/ : SUCCESS in 3m 55s\n- openstack-tox-functional http://logs.openstack.org/61/654261/1/check/openstack-tox-functional/f0f41f8/ : SUCCESS in 4m 27s\n- openstack-tox-functional-py36 http://logs.openstack.org/61/654261/1/check/openstack-tox-functional-py36/9693010/ : FAILURE in 2m 10s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/61/654261/1/check/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/001ff74/ : SUCCESS in 2h 26m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a124f77e3af78cbadc2b7ce50e32861a2ab51cc1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-04-22 18:00:16.000000000","message":"Patch Set 1:\n\nHi Abhiram, I think the problem you have is solved in [1]. If [1] is correct (let the reviewers first to see the patch), then rebase yours on top of this one.\n\nWelcome!!\n\n[1] https://review.opendev.org/#/c/654585/","accounts_in_message":[],"_revision_number":1},{"id":"3b121342d57d95dc414898bc33457ac5482a3315","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-04-23 07:22:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e46272b19526897efd87f5ad4edae64d5f84e918","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-24 09:06:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85344d6225f3be69bb2f32777f0142b49d4a4b3e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-04-24 09:10:25.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks for the functional test :) One comment inline","accounts_in_message":[],"_revision_number":2},{"id":"6ddd4999efe91c64e2841496f532bc58142401a6","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-24 10:16:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d50f38a0f5f90b1d31d3eab7454c85081c1465bd","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-24 11:34:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5c0d530a3e859085e470e23f2f8a9e435ceeb51f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-04-24 11:38:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1789516676896446c8ef4e43283ae68073b97695","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-24 11:44:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ddeeaa76f3a4f33fda22f31262db09758c910416","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-24 11:45:22.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c05260f43c59a09a365a832406202b250aa739d2","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-24 11:46:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3f1fab4fee9cdbc677db25972d856e30c9de5bac","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-24 11:46:39.000000000","message":"Patch Set 7:\n\nSorry about the multiple patches.","accounts_in_message":[],"_revision_number":7},{"id":"424bf4338f914e8b1ad0034f2a91aaafd813f526","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-24 11:48:15.000000000","message":"Patch Set 7:\n\nHow do I trigger the Zuul tests?","accounts_in_message":[],"_revision_number":7},{"id":"29fa10d8bcd76f6f1fec0b9cddeaa5e1ec961a17","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-04-24 11:53:00.000000000","message":"Patch Set 7:\n\n\u003e How do I trigger the Zuul tests?\n\nIt\u0027s triggered automatically. You can watch it e.g. here: http://zuul.openstack.org/stream/a71010f822e24e19af01093d0b40da33?logfile\u003dconsole.log\n\nif you search for change number 654261 (taken from URL) in http://zuul.openstack.org/status, you should find all the triggered jobs.","accounts_in_message":[],"_revision_number":7},{"id":"fdff3e4cbf16a4809484819782af182e1396bef1","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-24 11:53:56.000000000","message":"Patch Set 7:\n\nThank you","accounts_in_message":[],"_revision_number":7},{"id":"bea558b3a489b8bebf107c572f0f1bc342f8d3c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 14:27:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/61/654261/7/check/openstack-tox-pep8/85b29bb/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/61/654261/7/check/openstack-tox-py27/08a400f/ : SUCCESS in 3m 15s\n- openstack-tox-py35 http://logs.openstack.org/61/654261/7/check/openstack-tox-py35/60c8d0e/ : SUCCESS in 2m 51s\n- openstack-tox-py36 http://logs.openstack.org/61/654261/7/check/openstack-tox-py36/ba5489a/ : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/654261/7/check/openstack-tox-lower-constraints/cdfdd6b/ : SUCCESS in 3m 51s\n- openstack-tox-functional http://logs.openstack.org/61/654261/7/check/openstack-tox-functional/a71010f/ : FAILURE in 4m 32s\n- openstack-tox-functional-py36 http://logs.openstack.org/61/654261/7/check/openstack-tox-functional-py36/2ff5af9/ : FAILURE in 2m 20s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/61/654261/7/check/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/1d590ff/ : SUCCESS in 2h 30m 33s","accounts_in_message":[],"_revision_number":7},{"id":"e67ad5ba0808f70d448f9b609159bcbfab08c8e8","author":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"date":"2019-04-27 12:47:42.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"59052c40e00523df7c10ec75302f6e47b4404df5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-27 14:52:21.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/61/654261/8/check/openstack-tox-pep8/16eeccd/ : SUCCESS in 3m 12s\n- openstack-tox-py27 http://logs.openstack.org/61/654261/8/check/openstack-tox-py27/c1329af/ : SUCCESS in 2m 35s\n- openstack-tox-py35 http://logs.openstack.org/61/654261/8/check/openstack-tox-py35/cab9f4d/ : SUCCESS in 2m 44s\n- openstack-tox-py36 http://logs.openstack.org/61/654261/8/check/openstack-tox-py36/1f024c6/ : SUCCESS in 2m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/654261/8/check/openstack-tox-lower-constraints/f500888/ : SUCCESS in 4m 21s\n- openstack-tox-functional http://logs.openstack.org/61/654261/8/check/openstack-tox-functional/da5573c/ : SUCCESS in 3m 46s\n- openstack-tox-functional-py36 http://logs.openstack.org/61/654261/8/check/openstack-tox-functional-py36/0ab770d/ : FAILURE in 2m 16s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/61/654261/8/check/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/d3987ed/ : SUCCESS in 2h 03m 37s","accounts_in_message":[],"_revision_number":8},{"id":"9712cb3f52d6e79f36d1d29608bf455a2bdb8c43","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-05-02 15:57:25.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"3c8881df68e8d2db95f86f8c2fa22093d30240b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 18:59:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/61/654261/8/check/openstack-tox-pep8/d9424c2/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/61/654261/8/check/openstack-tox-py27/1c4427a/ : SUCCESS in 3m 06s\n- openstack-tox-py35 http://logs.openstack.org/61/654261/8/check/openstack-tox-py35/2dcec47/ : SUCCESS in 2m 52s\n- openstack-tox-py36 http://logs.openstack.org/61/654261/8/check/openstack-tox-py36/6815645/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/654261/8/check/openstack-tox-lower-constraints/7e38f5a/ : SUCCESS in 4m 47s\n- openstack-tox-functional http://logs.openstack.org/61/654261/8/check/openstack-tox-functional/f5c1e70/ : SUCCESS in 4m 12s\n- openstack-tox-functional-py36 http://logs.openstack.org/61/654261/8/check/openstack-tox-functional-py36/979454d/ : SUCCESS in 4m 37s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/61/654261/8/check/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/18ce5c8/ : SUCCESS in 2h 43m 36s","accounts_in_message":[],"_revision_number":8},{"id":"2445eaec19f2a4121d430423876cd3869c63bc50","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-05-31 19:44:48.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0cf077b088804356bd83c9c8dbf79887e2eee22f","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-05-31 19:52:59.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"41b4e6f19ea1645944dfdc286c6e28c8bf5a5f0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 19:53:11.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"158fe65924f59e5e424263e01f7d727f0267d708","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 22:11:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"2d44083b4c2021ba47a36dd708555bf9b4010b5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 22:11:47.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/61/654261/8/gate/openstack-tox-pep8/d9b8033/ : SUCCESS in 3m 57s\n- openstack-tox-py27 http://logs.openstack.org/61/654261/8/gate/openstack-tox-py27/321d916/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/61/654261/8/gate/openstack-tox-py35/762dcc5/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/61/654261/8/gate/openstack-tox-py36/198d486/ : SUCCESS in 3m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/61/654261/8/gate/openstack-tox-lower-constraints/8f05d29/ : SUCCESS in 3m 39s\n- openstack-tox-functional http://logs.openstack.org/61/654261/8/gate/openstack-tox-functional/45c5d29/ : SUCCESS in 4m 47s\n- openstack-tox-functional-py36 http://logs.openstack.org/61/654261/8/gate/openstack-tox-functional-py36/043f5a3/ : SUCCESS in 4m 49s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/61/654261/8/gate/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/048d912/ : SUCCESS in 2h 13m 10s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"9f8dc4494dd5b273a282d21cc15057086624029d","revisions":{"6a0f1d87030c209e90795b6e5202e41e3f43ca48":{"kind":"REWORK","_number":1,"created":"2019-04-21 19:01:21.000000000","uploader":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"ref":"refs/changes/61/654261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/61/654261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/1"}}},"commit":{"parents":[{"commit":"354c87d4b95511d7cd1b90d1d2737eabc4baa114","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/354c87d4b95511d7cd1b90d1d2737eabc4baa114"}]}],"author":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-21 18:36:32.000000000","tz":-420},"committer":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-21 18:56:31.000000000","tz":-420},"subject":"Allow match_ip in LrNatDelCommand to be IP network","message":"Allow match_ip in LrNatDelCommand to be IP network\n\n__init__ of LrNatDelCommand in schema/ovn_northbound/commands.py takes\noptional match_ip parameter to find one or more specific NAT Table rows\nof a logical router to delete. This match_ip is searched in the\n\"logical_ip\" column for NAT rows of type \"snat\" and in \"external_ip\"\ncolumn for other types. While \"external_ip\" column value has to be an\nIP address, \"logical_ip\" column value can be IPv4 network or\nIP address. Currently the code only allows match_ip to be an IP address\nand raises an error when an IP network is passed.\n\nThis change allows match_ip to be IP network when NAT type is \"snat\".\n\nCloses-Bug: #1825740\nChange-Id: I90dfdb5b5fe625ca8af768dfe9dac10f6d3c1430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6a0f1d87030c209e90795b6e5202e41e3f43ca48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6a0f1d87030c209e90795b6e5202e41e3f43ca48"}]},"branch":"refs/heads/master"},"93bede6a634de5e9040953ba0b9053ad3fdfe9d2":{"kind":"REWORK","_number":2,"created":"2019-04-24 09:06:41.000000000","uploader":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"ref":"refs/changes/61/654261/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/61/654261/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/2"}}},"commit":{"parents":[{"commit":"354c87d4b95511d7cd1b90d1d2737eabc4baa114","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/354c87d4b95511d7cd1b90d1d2737eabc4baa114"}]}],"author":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-21 18:36:32.000000000","tz":-420},"committer":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-24 09:04:42.000000000","tz":-420},"subject":"Allow match_ip in LrNatDelCommand to be IP network","message":"Allow match_ip in LrNatDelCommand to be IP network\n\n__init__ of LrNatDelCommand in schema/ovn_northbound/commands.py takes\noptional match_ip parameter to find one or more specific NAT Table rows\nof a logical router to delete. This match_ip is searched in the\n\"logical_ip\" column for NAT rows of type \"snat\" and in \"external_ip\"\ncolumn for other types. While \"external_ip\" column value has to be an\nIP address, \"logical_ip\" column value can be IPv4 network or\nIP address. Currently the code only allows match_ip to be an IP address\nand raises an error when an IP network is passed.\n\nThis change allows match_ip to be IP network when NAT type is \"snat\".\n\nCloses-Bug: #1825740\nChange-Id: I90dfdb5b5fe625ca8af768dfe9dac10f6d3c1430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/93bede6a634de5e9040953ba0b9053ad3fdfe9d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/93bede6a634de5e9040953ba0b9053ad3fdfe9d2"}]},"branch":"refs/heads/master"},"52e726edad30a5776eeed6024b396f31dc6dd0f6":{"kind":"REWORK","_number":3,"created":"2019-04-24 10:16:46.000000000","uploader":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"ref":"refs/changes/61/654261/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/61/654261/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/3"}}},"commit":{"parents":[{"commit":"354c87d4b95511d7cd1b90d1d2737eabc4baa114","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/354c87d4b95511d7cd1b90d1d2737eabc4baa114"}]}],"author":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-21 18:36:32.000000000","tz":-420},"committer":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-24 10:16:19.000000000","tz":-420},"subject":"Allow match_ip in LrNatDelCommand to be IP network","message":"Allow match_ip in LrNatDelCommand to be IP network\n\n__init__ of LrNatDelCommand in schema/ovn_northbound/commands.py takes\noptional match_ip parameter to find one or more specific NAT Table rows\nof a logical router to delete. This match_ip is searched in the\n\"logical_ip\" column for NAT rows of type \"snat\" and in \"external_ip\"\ncolumn for other types. While \"external_ip\" column value has to be an\nIP address, \"logical_ip\" column value can be IPv4 network or\nIP address. Currently the code only allows match_ip to be an IP address\nand raises an error when an IP network is passed.\n\nThis change allows match_ip to be IP network when NAT type is \"snat\".\n\nCloses-Bug: #1825740\nChange-Id: I90dfdb5b5fe625ca8af768dfe9dac10f6d3c1430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/52e726edad30a5776eeed6024b396f31dc6dd0f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/52e726edad30a5776eeed6024b396f31dc6dd0f6"}]},"branch":"refs/heads/master"},"ff141b16e609e7b6355e8f32a4c39b984cba3a80":{"kind":"REWORK","_number":4,"created":"2019-04-24 11:34:29.000000000","uploader":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"ref":"refs/changes/61/654261/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/61/654261/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/4"}}},"commit":{"parents":[{"commit":"354c87d4b95511d7cd1b90d1d2737eabc4baa114","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/354c87d4b95511d7cd1b90d1d2737eabc4baa114"}]}],"author":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-21 18:36:32.000000000","tz":-420},"committer":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-24 11:33:15.000000000","tz":-420},"subject":"Allow match_ip in LrNatDelCommand to be IP network","message":"Allow match_ip in LrNatDelCommand to be IP network\n\n__init__ of LrNatDelCommand in schema/ovn_northbound/commands.py takes\noptional match_ip parameter to find one or more specific NAT Table rows\nof a logical router to delete. This match_ip is searched in the\n\"logical_ip\" column for NAT rows of type \"snat\" and in \"external_ip\"\ncolumn for other types. While \"external_ip\" column value has to be an\nIP address, \"logical_ip\" column value can be IPv4 network or\nIP address. Currently the code only allows match_ip to be an IP address\nand raises an error when an IP network is passed.\n\nThis change allows match_ip to be IP network when NAT type is \"snat\".\n\nCloses-Bug: #1825740\nChange-Id: I90dfdb5b5fe625ca8af768dfe9dac10f6d3c1430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ff141b16e609e7b6355e8f32a4c39b984cba3a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ff141b16e609e7b6355e8f32a4c39b984cba3a80"}]},"branch":"refs/heads/master"},"3cbe2f97a079b0c1f947a6524a92e430132141cf":{"kind":"REWORK","_number":5,"created":"2019-04-24 11:44:34.000000000","uploader":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"ref":"refs/changes/61/654261/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/61/654261/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/5"}}},"commit":{"parents":[{"commit":"354c87d4b95511d7cd1b90d1d2737eabc4baa114","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/354c87d4b95511d7cd1b90d1d2737eabc4baa114"}]}],"author":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-21 18:36:32.000000000","tz":-420},"committer":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-24 11:44:15.000000000","tz":-420},"subject":"Allow match_ip in LrNatDelCommand to be IP network","message":"Allow match_ip in LrNatDelCommand to be IP network\n\n__init__ of LrNatDelCommand in schema/ovn_northbound/commands.py takes\noptional match_ip parameter to find one or more specific NAT Table rows\nof a logical router to delete. This match_ip is searched in the\n\"logical_ip\" column for NAT rows of type \"snat\" and in \"external_ip\"\ncolumn for other types. While \"external_ip\" column value has to be an\nIP address, \"logical_ip\" column value can be IPv4 network or\nIP address. Currently the code only allows match_ip to be an IP address\nand raises an error when an IP network is passed.\n\nThis change allows match_ip to be IP network when NAT type is \"snat\".\n\nCloses-Bug: #1825740\nChange-Id: I90dfdb5b5fe625ca8af768dfe9dac10f6d3c1430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3cbe2f97a079b0c1f947a6524a92e430132141cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3cbe2f97a079b0c1f947a6524a92e430132141cf"}]},"branch":"refs/heads/master"},"0b3fa97ab0e01fc7ff46b3440b3e89c74656b3ad":{"kind":"REWORK","_number":6,"created":"2019-04-24 11:45:22.000000000","uploader":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"ref":"refs/changes/61/654261/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/61/654261/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/6"}}},"commit":{"parents":[{"commit":"354c87d4b95511d7cd1b90d1d2737eabc4baa114","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/354c87d4b95511d7cd1b90d1d2737eabc4baa114"}]}],"author":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-21 18:36:32.000000000","tz":-420},"committer":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-24 11:45:18.000000000","tz":-420},"subject":"Allow match_ip in LrNatDelCommand to be IP network","message":"Allow match_ip in LrNatDelCommand to be IP network\n\n__init__ of LrNatDelCommand in schema/ovn_northbound/commands.py takes\noptional match_ip parameter to find one or more specific NAT Table rows\nof a logical router to delete. This match_ip is searched in the\n\"logical_ip\" column for NAT rows of type \"snat\" and in \"external_ip\"\ncolumn for other types. While \"external_ip\" column value has to be an\nIP address, \"logical_ip\" column value can be IPv4 network or\nIP address. Currently the code only allows match_ip to be an IP address\nand raises an error when an IP network is passed.\n\nThis change allows match_ip to be IP network when NAT type is \"snat\".\n\nCloses-Bug: #1825740\nChange-Id: I90dfdb5b5fe625ca8af768dfe9dac10f6d3c1430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0b3fa97ab0e01fc7ff46b3440b3e89c74656b3ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0b3fa97ab0e01fc7ff46b3440b3e89c74656b3ad"}]},"branch":"refs/heads/master"},"7cbacc77dee2ce956f2595c2b4e27c2b1757fcde":{"kind":"REWORK","_number":7,"created":"2019-04-24 11:46:18.000000000","uploader":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"ref":"refs/changes/61/654261/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/61/654261/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/7"}}},"commit":{"parents":[{"commit":"354c87d4b95511d7cd1b90d1d2737eabc4baa114","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/354c87d4b95511d7cd1b90d1d2737eabc4baa114"}]}],"author":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-21 18:36:32.000000000","tz":-420},"committer":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-24 11:46:14.000000000","tz":-420},"subject":"Allow match_ip in LrNatDelCommand to be IP network","message":"Allow match_ip in LrNatDelCommand to be IP network\n\n__init__ of LrNatDelCommand in schema/ovn_northbound/commands.py takes\noptional match_ip parameter to find one or more specific NAT Table rows\nof a logical router to delete. This match_ip is searched in the\n\"logical_ip\" column for NAT rows of type \"snat\" and in \"external_ip\"\ncolumn for other types. While \"external_ip\" column value has to be an\nIP address, \"logical_ip\" column value can be IPv4 network or\nIP address. Currently the code only allows match_ip to be an IP address\nand raises an error when an IP network is passed.\n\nThis change allows match_ip to be IP network when NAT type is \"snat\".\n\nCloses-Bug: #1825740\nChange-Id: I90dfdb5b5fe625ca8af768dfe9dac10f6d3c1430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7cbacc77dee2ce956f2595c2b4e27c2b1757fcde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7cbacc77dee2ce956f2595c2b4e27c2b1757fcde"}]},"branch":"refs/heads/master"},"9f8dc4494dd5b273a282d21cc15057086624029d":{"kind":"REWORK","_number":8,"created":"2019-04-27 12:47:42.000000000","uploader":{"_account_id":30297,"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","username":"Abh1ram"},"ref":"refs/changes/61/654261/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/61/654261/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/61/654261/8"}}},"commit":{"parents":[{"commit":"354c87d4b95511d7cd1b90d1d2737eabc4baa114","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/354c87d4b95511d7cd1b90d1d2737eabc4baa114"}]}],"author":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-21 18:36:32.000000000","tz":-420},"committer":{"name":"Abhiram Sangana","email":"sangana.abhiram@nutanix.com","date":"2019-04-27 12:45:06.000000000","tz":-420},"subject":"Allow match_ip in LrNatDelCommand to be IP network","message":"Allow match_ip in LrNatDelCommand to be IP network\n\n__init__ of LrNatDelCommand in schema/ovn_northbound/commands.py takes\noptional match_ip parameter to find one or more specific NAT Table rows\nof a logical router to delete. This match_ip is searched in the\n\"logical_ip\" column for NAT rows of type \"snat\" and in \"external_ip\"\ncolumn for other types. While \"external_ip\" column value has to be an\nIP address, \"logical_ip\" column value can be IPv4 network or\nIP address. Currently the code only allows match_ip to be an IP address\nand raises an error when an IP network is passed.\n\nThis change allows match_ip to be IP network when NAT type is \"snat\".\n\nCloses-Bug: #1825740\nChange-Id: I90dfdb5b5fe625ca8af768dfe9dac10f6d3c1430\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9f8dc4494dd5b273a282d21cc15057086624029d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9f8dc4494dd5b273a282d21cc15057086624029d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
