)]}'
{"id":"openstack%2Fovsdbapp~655070","triplet_id":"openstack%2Fovsdbapp~master~I447fa8594a7f11d5d8ed455353020b6871f596e5","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"I447fa8594a7f11d5d8ed455353020b6871f596e5","subject":"Update the UPPER_CONSTRAINTS_FILE for tox","status":"MERGED","created":"2019-04-23 11:40:15.000000000","updated":"2019-07-29 21:15:53.000000000","submitted":"2019-07-29 21:15:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"655070-1564434953240-c33df9d0","meta_rev_id":"1ba5fba755019f66ba24132875b8fb774c82dbed","_number":655070,"virtual_id_number":655070,"owner":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2019-07-29 21:15:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"recommended":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"all":[{"value":1,"date":"2019-07-05 05:42:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-06-27 13:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-07-29 20:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-07-29 20:58:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-24 03:44:14.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-06-27 13:27:48.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-07-05 05:42:48.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-07-29 20:58:08.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-07-29 21:15:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c66a14d0314b03ab54d5bc23c24e7d4546fe3b9","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 11:40:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c782202955cf64640e8faf858544175d652d3ca4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 12:28:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/70/655070/1/check/test-release-openstack/56f973b/ : SUCCESS in 2m 25s\n- openstack-tox-pep8 http://logs.openstack.org/70/655070/1/check/openstack-tox-pep8/0ea9b99/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/70/655070/1/check/openstack-tox-py27/ce6b8bf/ : SUCCESS in 2m 49s\n- openstack-tox-py35 http://logs.openstack.org/70/655070/1/check/openstack-tox-py35/f604152/ : SUCCESS in 5m 31s\n- openstack-tox-py36 http://logs.openstack.org/70/655070/1/check/openstack-tox-py36/3c7ce6a/ : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/655070/1/check/openstack-tox-lower-constraints/79f5e7f/ : SUCCESS in 5m 28s\n- openstack-tox-functional http://logs.openstack.org/70/655070/1/check/openstack-tox-functional/f93b2ae/ : SUCCESS in 4m 32s\n- openstack-tox-functional-py36 http://logs.openstack.org/70/655070/1/check/openstack-tox-functional-py36/0fb38e0/ : FAILURE in 2m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6fa289d759e3998661911ffc5ea086e804c12b49","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-04-24 03:44:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cd402ffd89caf5e7e59bce496344ec7df7a8e55","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-06-27 12:36:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc9eb6b6b738c45677d5fb658833166c0ee776c9","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-06-27 12:41:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0fce8e356fd60a4d7d0be259edec22b8de3382af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:25:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/70/655070/3/check/test-release-openstack/9a50b8e/ : SUCCESS in 2m 15s\n- openstack-tox-pep8 http://logs.openstack.org/70/655070/3/check/openstack-tox-pep8/87c9fb8/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/70/655070/3/check/openstack-tox-py27/a3add26/ : SUCCESS in 2m 39s\n- openstack-tox-py35 http://logs.openstack.org/70/655070/3/check/openstack-tox-py35/1a005b3/ : SUCCESS in 3m 57s\n- openstack-tox-py36 http://logs.openstack.org/70/655070/3/check/openstack-tox-py36/e7a22d5/ : SUCCESS in 2m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/655070/3/check/openstack-tox-lower-constraints/c54f0b5/ : SUCCESS in 4m 23s\n- openstack-tox-functional http://logs.openstack.org/70/655070/3/check/openstack-tox-functional/04a1b3b/ : SUCCESS in 4m 30s\n- openstack-tox-functional-py36 http://logs.openstack.org/70/655070/3/check/openstack-tox-functional-py36/2bcbd6d/ : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"e9f1e68a36027b70077f2d203120905286f7df71","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-06-27 13:27:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"838c9715f687578d5a54b4b814bce38b0955af96","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-07-05 05:42:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"13bd01bb331425339b8b088d2be0e094d2f02612","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-07-29 20:58:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"05bc4985e79322a75e0ed073e28de0d1f17ac68a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 20:58:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"52959946e9836004a52ed2e87a8ae195d2014880","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:15:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/70/655070/3/gate/test-release-openstack/b5602f2/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/70/655070/3/gate/openstack-tox-pep8/3782c8a/ : SUCCESS in 4m 43s\n- openstack-tox-py27 http://logs.openstack.org/70/655070/3/gate/openstack-tox-py27/59ada84/ : SUCCESS in 4m 27s\n- openstack-tox-py35 http://logs.openstack.org/70/655070/3/gate/openstack-tox-py35/9efda25/ : SUCCESS in 11m 44s\n- openstack-tox-py36 http://logs.openstack.org/70/655070/3/gate/openstack-tox-py36/82444f9/ : SUCCESS in 3m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/655070/3/gate/openstack-tox-lower-constraints/8eccf93/ : SUCCESS in 4m 34s\n- openstack-tox-functional http://logs.openstack.org/70/655070/3/gate/openstack-tox-functional/9b95bae/ : SUCCESS in 5m 16s\n- openstack-tox-functional-py36 http://logs.openstack.org/70/655070/3/gate/openstack-tox-functional-py36/702b89e/ : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":3},{"id":"1ba5fba755019f66ba24132875b8fb774c82dbed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:15:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bcafa4bf3ef8c99c4f8b844998fea88bab1b7f57","revisions":{"c3c60790fc7342c23344a2f43b22ca66fcc4d14c":{"kind":"REWORK","_number":1,"created":"2019-04-23 11:40:15.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/70/655070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/655070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/1"}}},"commit":{"parents":[{"commit":"354c87d4b95511d7cd1b90d1d2737eabc4baa114","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/354c87d4b95511d7cd1b90d1d2737eabc4baa114"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:40:12.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I447fa8594a7f11d5d8ed455353020b6871f596e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c3c60790fc7342c23344a2f43b22ca66fcc4d14c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c3c60790fc7342c23344a2f43b22ca66fcc4d14c"}]},"branch":"refs/heads/master"},"285ace3029d86da317786f1af4285f919a8ac416":{"kind":"REWORK","_number":2,"created":"2019-06-27 12:36:15.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/70/655070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/655070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/2"}}},"commit":{"parents":[{"commit":"a43f32a6a68d7fe56450913fa5f8300e23fa4e2b","subject":"Merge \"Move WaitForPortBindingEvent out from testing code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a43f32a6a68d7fe56450913fa5f8300e23fa4e2b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-06-27 12:36:06.000000000","tz":480},"subject":"Replace git.openstack.org URLs with opendev.org URLs","message":"Replace git.openstack.org URLs with opendev.org URLs\n\nChange-Id: I447fa8594a7f11d5d8ed455353020b6871f596e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/285ace3029d86da317786f1af4285f919a8ac416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/285ace3029d86da317786f1af4285f919a8ac416"}]},"branch":"refs/heads/master"},"bcafa4bf3ef8c99c4f8b844998fea88bab1b7f57":{"kind":"REWORK","_number":3,"created":"2019-06-27 12:41:42.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/70/655070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/655070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/655070/3"}}},"commit":{"parents":[{"commit":"a43f32a6a68d7fe56450913fa5f8300e23fa4e2b","subject":"Merge \"Move WaitForPortBindingEvent out from testing code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a43f32a6a68d7fe56450913fa5f8300e23fa4e2b"}]}],"author":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 11:39:46.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-06-27 12:40:14.000000000","tz":480},"subject":"Update the UPPER_CONSTRAINTS_FILE for tox","message":"Update the UPPER_CONSTRAINTS_FILE for tox\n\n1. Update the UPPER_CONSTRAINTS_FILE for tox[1]\n2. Remove unncessary \"\u003d\u003d\"\n3. Update some URLs to latest\n\n[1]: http://lists.openstack.org/pipermail/openstack-discuss/2019-May/006478.html\n\nChange-Id: I447fa8594a7f11d5d8ed455353020b6871f596e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bcafa4bf3ef8c99c4f8b844998fea88bab1b7f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bcafa4bf3ef8c99c4f8b844998fea88bab1b7f57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
