)]}'
{"id":"openstack%2Fovsdbapp~658428","triplet_id":"openstack%2Fovsdbapp~master~I55008aff94ec91ef87ea541b7040aacb30a73c66","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"I55008aff94ec91ef87ea541b7040aacb30a73c66","subject":"TrivialFix: Fix the chassis_del comment in Southbound API","status":"MERGED","created":"2019-05-10 17:16:21.000000000","updated":"2019-05-10 22:20:47.000000000","submitted":"2019-05-10 22:20:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658428-1557526847622-7f1a7d1e","meta_rev_id":"85d89431bc565d42c33de2e3b993b4804451af1d","_number":658428,"virtual_id_number":658428,"owner":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2019-05-10 22:20:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":2,"date":"2019-05-10 18:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":1,"date":"2019-05-10 18:13:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 18:13:38.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-05-10 22:20:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac3bdd050d48e6c4ecfc0994e50b7f21540f3517","author":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"date":"2019-05-10 17:16:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a52d807d9361dddb05bd281adacce381637089c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-05-10 18:13:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"096f04ff095747badb45d26ef19b9284ce07745c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:44:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/28/658428/1/check/openstack-tox-pep8/babd6ba/ : SUCCESS in 7m 52s\n- openstack-tox-py27 http://logs.openstack.org/28/658428/1/check/openstack-tox-py27/f83f305/ : SUCCESS in 3m 56s\n- openstack-tox-py35 http://logs.openstack.org/28/658428/1/check/openstack-tox-py35/4b4876b/ : SUCCESS in 3m 32s\n- openstack-tox-py36 http://logs.openstack.org/28/658428/1/check/openstack-tox-py36/999132d/ : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints http://logs.openstack.org/28/658428/1/check/openstack-tox-lower-constraints/140875c/ : SUCCESS in 4m 03s\n- openstack-tox-functional http://logs.openstack.org/28/658428/1/check/openstack-tox-functional/2b45032/ : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 http://logs.openstack.org/28/658428/1/check/openstack-tox-functional-py36/5e591db/ : SUCCESS in 5m 36s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/28/658428/1/check/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/d33539b/ : SUCCESS in 2h 21m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5260baa8e74d345c77ad83a19a496e7c7bc70e0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 19:44:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77143524b79503867671c88aa44390a8ca3361c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 22:20:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"85d89431bc565d42c33de2e3b993b4804451af1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 22:20:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/28/658428/1/gate/openstack-tox-pep8/320ca61/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/28/658428/1/gate/openstack-tox-py27/b26669c/ : SUCCESS in 2m 56s\n- openstack-tox-py35 http://logs.openstack.org/28/658428/1/gate/openstack-tox-py35/b190d6e/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/28/658428/1/gate/openstack-tox-py36/2bfcd0d/ : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/28/658428/1/gate/openstack-tox-lower-constraints/b027a46/ : SUCCESS in 4m 12s\n- openstack-tox-functional http://logs.openstack.org/28/658428/1/gate/openstack-tox-functional/1dc909e/ : SUCCESS in 5m 21s\n- openstack-tox-functional-py36 http://logs.openstack.org/28/658428/1/gate/openstack-tox-functional-py36/dadf270/ : SUCCESS in 4m 17s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/28/658428/1/gate/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/1a85056/ : SUCCESS in 2h 32m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fe5351c09d579e36d434a9b5529ce87629e0e25","revisions":{"0fe5351c09d579e36d434a9b5529ce87629e0e25":{"kind":"REWORK","_number":1,"created":"2019-05-10 17:16:21.000000000","uploader":{"_account_id":17776,"name":"Reedip","email":"reedip.banerjee@gmail.com","username":"Reedip"},"ref":"refs/changes/28/658428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/28/658428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/658428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/658428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/28/658428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/28/658428/1"}}},"commit":{"parents":[{"commit":"048473da01283fcd18c246b7298da1a4a34df504","subject":"Add missing tox environment \"functional-py36\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/048473da01283fcd18c246b7298da1a4a34df504"}]}],"author":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-05-10 17:15:42.000000000","tz":0},"committer":{"name":"reedip","email":"rbanerje@redhat.com","date":"2019-05-10 17:16:20.000000000","tz":0},"subject":"TrivialFix: Fix the chassis_del comment in Southbound API","message":"TrivialFix: Fix the chassis_del comment in Southbound API\n\nChange-Id: I55008aff94ec91ef87ea541b7040aacb30a73c66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0fe5351c09d579e36d434a9b5529ce87629e0e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0fe5351c09d579e36d434a9b5529ce87629e0e25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
