)]}'
{"id":"openstack%2Fovsdbapp~669435","triplet_id":"openstack%2Fovsdbapp~master~I1c658ddb1e4470dcf1f547747c9f90fdb08ad438","project":"openstack/ovsdbapp","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I1c658ddb1e4470dcf1f547747c9f90fdb08ad438","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 17:56:17.000000000","updated":"2019-07-29 23:28:26.000000000","submitted":"2019-07-29 23:28:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669435-1564442906140-a91269ad","meta_rev_id":"ce791d2d1ff9f0f8ee6f281f6c95648e916027c1","_number":669435,"virtual_id_number":669435,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2019-07-29 23:28:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-07-26 20:25:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":1,"date":"2019-07-27 12:34:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2019-07-29 20:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"date":"2019-07-26 20:25:27.000000000","_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2019-07-29 20:58:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 20:25:04.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-07-26 20:25:27.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-07-27 12:34:28.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-07-29 20:58:57.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-07-29 23:28:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"001ef8db65dc2d5d143890f7a46cba8ddee21f7a","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 17:56:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d6a06ec0a5c822b3b9d1590e055ce863f7626a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 20:33:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/35/669435/1/check/test-release-openstack/49949cf/ : SUCCESS in 2m 04s\n- openstack-tox-pep8 http://logs.openstack.org/35/669435/1/check/openstack-tox-pep8/6e9b11c/ : SUCCESS in 4m 13s\n- openstack-tox-py27 http://logs.openstack.org/35/669435/1/check/openstack-tox-py27/c20b4af/ : SUCCESS in 4m 21s\n- openstack-tox-py36 http://logs.openstack.org/35/669435/1/check/openstack-tox-py36/fefe114/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/35/669435/1/check/openstack-tox-py37/1471705/ : SUCCESS in 3m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/669435/1/check/openstack-tox-lower-constraints/21195e8/ : SUCCESS in 4m 39s\n- openstack-tox-functional http://logs.openstack.org/35/669435/1/check/openstack-tox-functional/e2419cb/ : SUCCESS in 4m 30s\n- openstack-tox-functional-py36 http://logs.openstack.org/35/669435/1/check/openstack-tox-functional-py36/eee93a6/ : SUCCESS in 4m 13s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/35/669435/1/check/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/5c53977/ : SUCCESS in 2h 32m 46s","accounts_in_message":[],"_revision_number":1},{"id":"5539db688529d8269bcefdd453c90dc09304dd59","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-07-26 20:25:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f6991ef7fb516b8e03160e22b35210826ac2b23e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-07-27 12:34:28.000000000","message":"Patch Set 1: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"1e21b67e5804d2d000059cdaf430a268c1964121","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-07-29 20:58:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9245450d2455a43ccaf476bb2260ede260b6b479","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 20:59:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69bbf747925a83d17350be1588fee9dddd4f55ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 23:28:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/35/669435/1/gate/test-release-openstack/5d465b6/ : SUCCESS in 4m 13s\n- openstack-tox-pep8 http://logs.openstack.org/35/669435/1/gate/openstack-tox-pep8/9ce9d02/ : SUCCESS in 4m 37s\n- openstack-tox-py27 http://logs.openstack.org/35/669435/1/gate/openstack-tox-py27/4b347d2/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/35/669435/1/gate/openstack-tox-py36/cf04953/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/35/669435/1/gate/openstack-tox-py37/2fe321f/ : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/35/669435/1/gate/openstack-tox-lower-constraints/430e114/ : SUCCESS in 5m 15s\n- openstack-tox-functional http://logs.openstack.org/35/669435/1/gate/openstack-tox-functional/98a640e/ : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 http://logs.openstack.org/35/669435/1/gate/openstack-tox-functional-py36/a500601/ : SUCCESS in 5m 28s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/35/669435/1/gate/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/6d6c468/ : SUCCESS in 2h 12m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c750e641adcd5d8ea106285a505f38eeb842ea94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 23:28:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a9b9e22f2f876d2e14604e70dfccca1c5775f59","revisions":{"5a9b9e22f2f876d2e14604e70dfccca1c5775f59":{"kind":"REWORK","_number":1,"created":"2019-07-05 17:56:17.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/35/669435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/35/669435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/35/669435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/35/669435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/35/669435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/35/669435/1"}}},"commit":{"parents":[{"commit":"81c124fe291e3156473ec9ffdd1087202c4974c8","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/81c124fe291e3156473ec9ffdd1087202c4974c8"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:46:02.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 17:56:16.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I1c658ddb1e4470dcf1f547747c9f90fdb08ad438\nStory: #2005924\nTask: #34225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5a9b9e22f2f876d2e14604e70dfccca1c5775f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5a9b9e22f2f876d2e14604e70dfccca1c5775f59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
