)]}'
{"id":"openstack%2Fovsdbapp~671765","triplet_id":"openstack%2Fovsdbapp~master~I0a8e54b0e61553e5962ccc8030a2e8e9a9d49390","project":"openstack/ovsdbapp","branch":"master","topic":"add-lrp-get-set-options","hashtags":[],"change_id":"I0a8e54b0e61553e5962ccc8030a2e8e9a9d49390","subject":"Add commands to set and get LRP options","status":"MERGED","created":"2019-07-19 16:02:22.000000000","updated":"2019-08-05 13:05:49.000000000","submitted":"2019-08-05 13:05:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671765-1565010349345-b796e31c","meta_rev_id":"835682663ab1c5dc5718e1309c38ef346bcc5ee8","_number":671765,"virtual_id_number":671765,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":2,"date":"2019-08-05 13:05:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"recommended":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"all":[{"value":1,"date":"2019-08-01 09:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":2,"date":"2019-07-29 21:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2019-08-05 10:57:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2019-08-05 10:57:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 21:00:12.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-08-01 09:09:31.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2019-08-05 10:57:22.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-08-05 13:05:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53ff13a94f601be2f3d88c51b1874aace97791f6","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-07-19 16:02:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4818b1d51f7f634c2c34b33ce0a7dc3d25cd4db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:21:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/65/671765/1/check/openstack-tox-pep8/a9a9394/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/65/671765/1/check/openstack-tox-py27/057a413/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/65/671765/1/check/openstack-tox-py35/062373c/ : SUCCESS in 4m 29s\n- openstack-tox-py36 http://logs.openstack.org/65/671765/1/check/openstack-tox-py36/414c2a0/ : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/65/671765/1/check/openstack-tox-lower-constraints/59ba9ce/ : SUCCESS in 4m 24s\n- openstack-tox-functional http://logs.openstack.org/65/671765/1/check/openstack-tox-functional/12c5d34/ : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 http://logs.openstack.org/65/671765/1/check/openstack-tox-functional-py36/8ab8346/ : SUCCESS in 5m 29s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release http://logs.openstack.org/65/671765/1/check/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/d8f3bd6/ : SUCCESS in 2h 17m 32s","accounts_in_message":[],"_revision_number":1},{"id":"1405623a8dd0eeaac6b3d087f5c55a1294eb273a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-07-29 21:00:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8d7b670065e12ae9c98d5289838457d23dd72315","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2019-08-01 09:09:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"673951ed3262415c8fc2b8d4b6c0df24deb1221c","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-08-05 10:57:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"18186624fd03962cc60463fe98d71aed7ac14a6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 10:57:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d3f957ef9eb6214ea39083a05f84b83a085b493","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:05:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44fb5d5e6613fb04c0449e00ae15467c4fc51360","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:05:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/65/671765/1/gate/openstack-tox-pep8/7b35c8c/ : SUCCESS in 5m 26s\n- openstack-tox-py27 https://logs.opendev.org/65/671765/1/gate/openstack-tox-py27/b4ac81a/ : SUCCESS in 3m 40s\n- openstack-tox-py36 https://logs.opendev.org/65/671765/1/gate/openstack-tox-py36/7189f58/ : SUCCESS in 4m 40s\n- openstack-tox-py37 https://logs.opendev.org/65/671765/1/gate/openstack-tox-py37/622479a/ : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints https://logs.opendev.org/65/671765/1/gate/openstack-tox-lower-constraints/d3eff3a/ : SUCCESS in 5m 51s\n- openstack-tox-functional https://logs.opendev.org/65/671765/1/gate/openstack-tox-functional/bf5dffa/ : SUCCESS in 6m 51s\n- openstack-tox-functional-py36 https://logs.opendev.org/65/671765/1/gate/openstack-tox-functional-py36/f298826/ : SUCCESS in 4m 36s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://logs.opendev.org/65/671765/1/gate/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/0e7ddd0/ : SUCCESS in 2h 02m 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78de8a6187d3efc6f0712ece6642539d8c7a9f19","revisions":{"78de8a6187d3efc6f0712ece6642539d8c7a9f19":{"kind":"REWORK","_number":1,"created":"2019-07-19 16:02:22.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/65/671765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/65/671765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/65/671765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/65/671765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/65/671765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/65/671765/1"}}},"commit":{"parents":[{"commit":"81c124fe291e3156473ec9ffdd1087202c4974c8","subject":"Sync Sphinx requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/81c124fe291e3156473ec9ffdd1087202c4974c8"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-19 15:59:42.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2019-07-19 15:59:42.000000000","tz":60},"subject":"Add commands to set and get LRP options","message":"Add commands to set and get LRP options\n\nThis patch is adding two new commands:\n\n* lrp_set_options(): To set the \"options\" column for a LRP\n* lrp_get_options(): To get the values of the \"options\" column of a LRP\n\nSince the code is pretty much the same as the equivalent command for LSP\nthis patch attempts to reuse the same command for both.\n\nChange-Id: I0a8e54b0e61553e5962ccc8030a2e8e9a9d49390\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/78de8a6187d3efc6f0712ece6642539d8c7a9f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/78de8a6187d3efc6f0712ece6642539d8c7a9f19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
