)]}'
{"id":"openstack%2Fovsdbapp~674196","triplet_id":"openstack%2Fovsdbapp~master~Idaf347c183fe35d346f15b24c0bfad0585b082fc","project":"openstack/ovsdbapp","branch":"master","topic":"openstackdocstheme","hashtags":[],"change_id":"Idaf347c183fe35d346f15b24c0bfad0585b082fc","subject":"Bump the openstackdocstheme extension to 1.20","status":"ABANDONED","created":"2019-08-02 06:34:11.000000000","updated":"2019-10-08 01:32:09.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fb6c10c83c347426b9cfb607adcd83081f0bd3bc","_number":674196,"virtual_id_number":674196,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-04 19:57:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":-1,"date":"2019-10-05 03:03:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 08:46:36.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-09-10 17:06:56.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-10-04 19:57:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-05 03:03:08.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"ec6c11caa6f791a06a0882d0c756eb6ddf7b6d11","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-08-02 06:34:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56061edb3307b88af08ed21085c8a179034b083a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 09:01:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/96/674196/1/check/openstack-tox-pep8/ab68923/ : SUCCESS in 5m 14s\n- openstack-tox-py27 https://logs.opendev.org/96/674196/1/check/openstack-tox-py27/65b709a/ : SUCCESS in 4m 30s\n- openstack-tox-py36 https://logs.opendev.org/96/674196/1/check/openstack-tox-py36/a855f0c/ : SUCCESS in 4m 44s\n- openstack-tox-py37 https://logs.opendev.org/96/674196/1/check/openstack-tox-py37/6ff86a6/ : SUCCESS in 5m 12s\n- requirements-check https://logs.opendev.org/96/674196/1/check/requirements-check/b21e0bb/ : SUCCESS in 4m 04s\n- openstack-tox-lower-constraints https://logs.opendev.org/96/674196/1/check/openstack-tox-lower-constraints/d07cc66/ : SUCCESS in 6m 01s\n- openstack-tox-functional https://logs.opendev.org/96/674196/1/check/openstack-tox-functional/c3d997f/ : SUCCESS in 4m 58s\n- openstack-tox-functional-py36 https://logs.opendev.org/96/674196/1/check/openstack-tox-functional-py36/ca9d7c3/ : SUCCESS in 6m 10s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://logs.opendev.org/96/674196/1/check/ovsdbapp-tempest-dsvm-networking-ovn-ovs-release/c8d76aa/ : SUCCESS in 2h 20m 26s","accounts_in_message":[],"_revision_number":1},{"id":"285d15245121fa415a74dd0de3b95afe8de03409","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-08-22 08:46:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce5002503f3a9efa793984e9ce806ed218ca2dae","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-08-23 08:33:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"544f591dccc275e1a859dab57dff14c625737f4e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-10 17:06:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d500d6695f0a48cf52e1cd0e2a0231dc4584d625","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-10 17:06:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26bb630cf8c5f996044f926523cd7d43e337fae1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 19:09:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac5851b14a24c249d190103b820fdd6 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbef73ba652e4dbcb135188c30d7efff : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b7f5bcc5b564703b1f084429393a012 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2dffe4c41fc4815be80a6ba85a15f8d : SUCCESS in 6m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d80e76aaf0d3425590cf203044c3806e : SUCCESS in 2m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d3253ee5a0d4c6888215e60a444a024 : SUCCESS in 5m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/92d0edfd66fa409bb465c4f285d270c0 : FAILURE in 4m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0712c3185e1b4eceae9ce738284e77d9 : FAILURE in 5m 28s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/36d012d78b5042db94493f8256593bfa : SUCCESS in 1h 57m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4dc43bc503c8df6a30d3e4981ee278b3a0a506a7","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-09-29 01:51:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"23c75219b8dbd5ae20a4ccf2361292d19f716739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-29 04:28:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/da35843a37a4498eb2ded61e5b68eae2 : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da7345ab55cd457c9d7e737de77fe116 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e228a9fd02644ee9ba0f6d67ceb2652b : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/399d556a9bd14302ba3002a32c646344 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c35b67596294ea280bd41af06c8dec4 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf2a4e3ab90d49778a1ea2362c1e667d : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5df45050b444e278c0e4b8350bba43c : SUCCESS in 5m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c001cdd59a294ab1901f634612d1ad20 : FAILURE in 3m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d9af12642f304655930d522f09e8aba6 : SUCCESS in 6m 27s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/f87380afcb5b43e2b8a71a6225ee99de : SUCCESS in 2h 31m 07s","accounts_in_message":[],"_revision_number":3},{"id":"8e5c54923b05d48b766cbce862410f110f698fd4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 17:14:33.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"bffa76df4fa7af25b625f272f4bdf1e87c752a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 19:57:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/344b1e2a905c4b7abe28497c84ec3a37 : SUCCESS in 2m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f78d0d06e53472ebf00cb835d99c09e : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45e4349433e34c17b660d5e09d61caaf : SUCCESS in 5m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ffaa0a25aa8e4e2f8de589c9bb99163e : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cd28dac211e45e89e619d023cfc4990 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68977116b6064e2eb1dd5fefb1add8a8 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9661a8327a6462d97a1c7c1ff49fd9a : SUCCESS in 5m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a2b88232f3f8497b92c98a7ee4115989 : SUCCESS in 5m 53s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/0dffbb7e1cb6421b91b504306d81bb5e : SUCCESS in 2h 15m 04s","accounts_in_message":[],"_revision_number":4},{"id":"3ab5408fb2ae6978394eb85a936bb8cf4cbfd58e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-05 03:03:08.000000000","message":"Patch Set 4: Code-Review-1\n\nIt was already by commit ac3d3de9ecc3bd26c92f3480c067667771010d7e (as part of a patch to enable doc publishing).","accounts_in_message":[],"_revision_number":4},{"id":"f0230c30e576511daa2efdfc3ade877d65d87c30","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-08 01:32:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6597a96d89b4d0a4b7ebe8e149346cd0f08a3c74","revisions":{"c2c0c6bdb06051a2b1cf0b9ac7d4c26972bd254d":{"kind":"REWORK","_number":1,"created":"2019-08-02 06:34:11.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/96/674196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/96/674196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/1"}}},"commit":{"parents":[{"commit":"c3f1bd3465306cb87360658e47922455c500a6c3","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c3f1bd3465306cb87360658e47922455c500a6c3"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 06:34:11.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 06:34:14.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Idaf347c183fe35d346f15b24c0bfad0585b082fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c2c0c6bdb06051a2b1cf0b9ac7d4c26972bd254d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c2c0c6bdb06051a2b1cf0b9ac7d4c26972bd254d"}]},"branch":"refs/heads/master"},"085472af0eb850d6d928403511b34f7f2baf867d":{"kind":"REWORK","_number":2,"created":"2019-09-10 17:06:39.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/96/674196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/96/674196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/2"}}},"commit":{"parents":[{"commit":"c3f1bd3465306cb87360658e47922455c500a6c3","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c3f1bd3465306cb87360658e47922455c500a6c3"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 06:34:11.000000000","tz":480},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-10 17:06:35.000000000","tz":0},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Idaf347c183fe35d346f15b24c0bfad0585b082fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/085472af0eb850d6d928403511b34f7f2baf867d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/085472af0eb850d6d928403511b34f7f2baf867d"}]},"branch":"refs/heads/master"},"9c6ba38af3e3d2eb9733860c7a83f0fde1f01700":{"kind":"REWORK","_number":3,"created":"2019-09-29 01:51:50.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/96/674196/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/96/674196/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/3"}}},"commit":{"parents":[{"commit":"4310bd2a06ea60b48930bb7e94276365fcaeb4fc","subject":"Merge \"PDF documentation build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4310bd2a06ea60b48930bb7e94276365fcaeb4fc"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 06:34:11.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-09-29 01:52:28.000000000","tz":480},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Idaf347c183fe35d346f15b24c0bfad0585b082fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9c6ba38af3e3d2eb9733860c7a83f0fde1f01700"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9c6ba38af3e3d2eb9733860c7a83f0fde1f01700"}]},"branch":"refs/heads/master"},"6597a96d89b4d0a4b7ebe8e149346cd0f08a3c74":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-04 17:14:33.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/96/674196/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/96/674196/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/96/674196/4"}}},"commit":{"parents":[{"commit":"92627f4de9cdcc3b0f984e7fe7b54beb6329041b","subject":"Merge \"Fix building OVS after python2 removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/92627f4de9cdcc3b0f984e7fe7b54beb6329041b"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-08-02 06:34:11.000000000","tz":480},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 17:14:33.000000000","tz":0},"subject":"Bump the openstackdocstheme extension to 1.20","message":"Bump the openstackdocstheme extension to 1.20\n\nSome options are now automatically configured by the version 1.20:\n- project\n- html_last_updated_fmt\n- latex_engine\n- latex_elements\n- version\n- release.\n\nChange-Id: Idaf347c183fe35d346f15b24c0bfad0585b082fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6597a96d89b4d0a4b7ebe8e149346cd0f08a3c74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6597a96d89b4d0a4b7ebe8e149346cd0f08a3c74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
