)]}'
{"id":"openstack%2Fovsdbapp~678076","triplet_id":"openstack%2Fovsdbapp~master~I9c7994d50066adf02b0066cb8efb1beb61d2e1f0","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"I9c7994d50066adf02b0066cb8efb1beb61d2e1f0","subject":"Add indexed column support to ovsdbapp","status":"MERGED","created":"2019-08-22 17:52:22.000000000","updated":"2020-05-30 19:51:18.000000000","submitted":"2020-05-19 17:56:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":50,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678076-1589911015604-472850b6","meta_rev_id":"8941ca89e480311402e98efadaef09b3579858f6","_number":678076,"virtual_id_number":678076,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-19 17:56:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-18 14:48:55.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2020-05-15 17:16:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-05-18 14:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2020-05-19 15:29:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2020-05-19 15:29:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 13:35:12.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2020-05-15 06:55:14.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-05-15 07:52:16.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-05-15 17:16:28.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-05-18 14:07:04.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-05-19 15:29:28.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2020-05-19 17:56:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6de16cc9759b3e9940f0bd42ad261522d2ba4290","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-08-22 17:52:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"062d12ff1ab7892a298f9bbadbc177afcd36b24d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 18:29:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ce185152b2740949567d7063f75d59d : FAILURE in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/038975287a3b409b8d006733d57c5a37 : FAILURE in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/782de4244d1f46a4a041b73ca43db38d : FAILURE in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43e947fb0c8740d488fb610dc1ad0997 : FAILURE in 4m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3e3bf6ad6564465abe71f6508ab6668 : FAILURE in 5m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7f054f1c1e734bb7a99561e74220f7d4 : SUCCESS in 4m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9dd973e40b34c64950e8f773e7cbb47 : FAILURE in 7m 11s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/b87c4f16b42648ceaea3510475ae0c1b : FAILURE in 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d394469c00b7fcae5cbce2494635e2a3711d274f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-08-22 21:24:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5388ccc0407eafd6d2901746ca443e2c93ad0c76","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-08-22 22:12:51.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c438dceebbb8ab1e5249a40835134dd9787f78ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 22:14:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ce32e11c944c96846bcfe782b88c12 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7bbea0706b1497f86f6a1449e6cd4ad : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e3f00d6295d49bd85c983325adfb57a : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df3dbd73bb6f43fab206b15998162e44 : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a44a5b78162463494de8aaf29efd8fe : SUCCESS in 6m 45s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9697dd5ee3ac40bda148b8c65a037550 : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d926ff1c28b456db5001899a57be1e9 : SUCCESS in 5m 02s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/11990a9959b742408ab40abad0c23457 : FAILURE in 32m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d975c026b86a65933b81118a6edf2af22c38ab8b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-08-22 23:20:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f92ec3a5a04850edc172a2b762f611a9d870d64f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-08-23 01:42:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5e799d3d2e02ea94bfeae3f243fdc3283d9c2a9f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-08-23 03:57:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"234494401e05109a7823b65508983d08de6fa36a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 06:17:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/647eb9ac0ed646d49fd42df06722729c : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/788ca9f91cb84c95a1b2d350e118d124 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/367c2e8caa7d4462a1e960f45fbd3316 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d992fcfbe8f4a2cbf8e5e9dad66bb04 : SUCCESS in 3m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa2dc0f2e9be45e180e4af79d116ecac : SUCCESS in 4m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/74958352629d408c9081eb1b10d3779e : SUCCESS in 5m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/29d4ea6770954754bdc6130c26adfdd4 : SUCCESS in 5m 17s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/698ebd9ee5c345c6b9078fe95e5b1da0 : SUCCESS in 2h 17m 45s","accounts_in_message":[],"_revision_number":4},{"id":"5d4e2e610b9848e0d4c8738ae0b8b870f57a6040","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-12 16:03:22.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"61d2748628b505b7da9b1873d95e15f6c6e4dba3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-12 16:14:43.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"9e2aa56b5af3e4d48e0901a1c53f02aa0c002de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 18:01:21.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f28cabbe2dde483db64a9aa18481dbef : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea4f002420d4870bf33779a2a746097 : FAILURE in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd4fc17a238d4dfd9dba3c6dbe2286ac : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/359c097ad615472692467f2c668f2153 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b7a762876824dc0965a4319a25048a8 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97d5300105b14660af62a4e0407f0349 : SUCCESS in 5m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/deb1c03e7f054ff28a040e5d4abcc702 : SUCCESS in 7m 49s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/07462cde164c4d18a600ed9b3aca1422 : FAILURE in 1h 43m 55s","accounts_in_message":[],"_revision_number":6},{"id":"7d59f00dc21943a36420b17f0353f989824fd83e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-12 21:25:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bc6399afbd0c25bf6408aa4d11b7a8d9e510457e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 23:31:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbd423bfde0b45009924ebdba31e9b23 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dafe7cfc2dc34f23b372e0364875afd8 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59481e259bcb4b5ca02ee9447ee3bd34 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1de85e5aa837495a9ac9e4aaa21b88f0 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb034a2b54de4a68a47a9a50402d597a : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b967064da24c768ee733871ced3dbc : SUCCESS in 7m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ac315d9a4873458dbe802f06b8e634b0 : SUCCESS in 7m 37s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d4c6ac89bac247949e4e466e17a44443 : SUCCESS in 1h 59m 50s","accounts_in_message":[],"_revision_number":7},{"id":"de840e61e551f14b4450e6175e7eb4ac943a9115","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-14 01:25:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"641be2d594e412641f583a11979003dfbf0ae154","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-14 01:45:58.000000000","message":"Patch Set 8:\n\nApparently other tests running in the same process don\u0027t always play nice with 5000 Logical Switch transactions being executed. I\u0027ll reduce to 2k which also fails w/o indexes, but should be a little easier on the system otherwise. May just have to remove that test, but I kind of wanted to show off since doing 20k only takes like 8 seconds. :p","accounts_in_message":[],"_revision_number":8},{"id":"a74b723601e9a97a351b1bb9a8484426bea7ddc9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-14 01:48:06.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"ddcf452734e0bb1b899ad015275b80299b212fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 04:11:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a22b50f5d5414aa1969d1227189cb762 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f86b9dafbb2d4fd8a0d03bfefb3c3055 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eabb2e2a7dd5429db304a34eb0c1e522 : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/816e79abe4e24f40ae01af024a17b7d5 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e742bd39d9d489da98c131ee473d3d4 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c532763e86849eda12678ddd07be712 : SUCCESS in 3m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0b23e206b38c44af9ff736310c89ba94 : SUCCESS in 5m 37s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/75b2d06e63824153af7078920d2767c1 : SUCCESS in 2h 18m 13s","accounts_in_message":[],"_revision_number":9},{"id":"b8a52073a56fa98d3458f0d411efd7dcd8ef2d2c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-05-14 07:41:31.000000000","message":"Patch Set 9: Code-Review+1\n\n(5 comments)\n\nGreat work!","accounts_in_message":[],"_revision_number":9},{"id":"62fad4120b94ebd46cecbf588cea34a1c0a2ea4c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-05-14 08:53:58.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)\n\nJust little questions, but looks goooood :)","accounts_in_message":[],"_revision_number":9},{"id":"65b46b2460840c1a7e761e9b82431bf66d57b1ff","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-05-14 13:42:26.000000000","message":"Patch Set 9: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4b18f32c70ee0f20cb5d3c5e6b5dc383e54e009d","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-14 19:35:14.000000000","message":"Patch Set 9:\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5fedfdbc27cb01f88006c0f6cd3b3036e449a1e9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-14 19:35:24.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"46aeee3a41ba87ce8ffa7ce15521291e25e468c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 21:53:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1257295ac2914102b8c9840945df8627 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb8da24279194571a4f74dc36464891e : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f4c300338b54ecfb6e3dd349e022723 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72f7735b6e824a5c8afff9a4b806b013 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db2c91dd6442428d99e1d676b763b619 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bdd615d80184daba01ff8049d8a0e5e : SUCCESS in 7m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cf4cb4e89096482eaac5cac9c05fe088 : SUCCESS in 6m 47s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d64c65a594c14acbaec2d4c9a18b4e84 : SUCCESS in 2h 16m 32s","accounts_in_message":[],"_revision_number":10},{"id":"e03fb2492572909cf7099cb4d919d58e66c74bcd","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-15 00:01:54.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"3ab0fa3d4ee2bee565a307dacff7a14926bab61e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-15 00:07:04.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"5d2d4092ad193ee08836f65726a225bd561a6438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 02:15:07.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/113bb65cd2c046a99c04da609e334a7f : FAILURE in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8b1a6403594b7db65b3ab07b6fb7d4 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/395757761a364f6c839e52eb14c4d491 : FAILURE in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e671530e7964354a0e16839baea7fd7 : FAILURE in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5087e544fbd54dd4a46f28002266ab80 : FAILURE in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae616b5e86ab491ebb6c25e573fa4ff9 : SUCCESS in 4m 53s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8507542fd9034c938080328aba65d9f1 : SUCCESS in 6m 08s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0ffc3eb1699b42e2ab885e10598b7d42 : SUCCESS in 2h 08m 14s","accounts_in_message":[],"_revision_number":11},{"id":"966cb49a5aa88076884377507995b538bec8f326","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-15 03:52:11.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"53e52dbeb4ffd327a3060801aa723281b8abe26b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-15 04:13:55.000000000","message":"Patch Set 12:\n\nrecheck\n\nran locally 100x w/o failure, not sure where that lsp_bind timeout is coming from in functional.","accounts_in_message":[],"_revision_number":12},{"id":"5d28c32546d5c2e30ede79dc5d6b7cc1a7c05ba4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-15 04:18:56.000000000","message":"Uploaded patch set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"0d25beb79ee6070c15589652b29c26b98524ed61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 06:42:17.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c7822d5e8704878aaad5a27ee28ead4 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c45cf5ff8244e18b39a33a52a5d3f25 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cb96137476e4788962b3548b56684b9 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/430c2638391c481e9c1ab35085325bac : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/503c2c92d9564fc7a666cb98d4fd70a8 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94b80d4bfdd740529ed9ebbdae1cf880 : SUCCESS in 4m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d107a3a265f744629e8c14b02fc78ae4 : SUCCESS in 5m 06s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/9220918587094697b292ebccddafaf6e : SUCCESS in 2h 18m 04s","accounts_in_message":[],"_revision_number":13},{"id":"7ea725a330bbc0ec58320897affd946816f2f383","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-05-15 06:55:14.000000000","message":"Patch Set 13: Code-Review+2","accounts_in_message":[],"_revision_number":13},{"id":"72a2c4cde4e01c2e186c2c4b2e30ba26e112432a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-05-15 07:51:23.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4b6c43c362a271321a5f64e67cb2ee06eb3a39c0","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-05-15 07:52:16.000000000","message":"Patch Set 13: Code-Review+1\n\nLGTM! Lets check the results and maybe create a comparison between browbeat runs with and without this patch! :)","accounts_in_message":[],"_revision_number":13},{"id":"425cd60261f3564ccb32e9b0411c2f8e2765e3be","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-05-15 10:30:17.000000000","message":"Patch Set 13:\n\n(5 comments)\n\nI got some questions/nits -- as usual -- but lgtm!","accounts_in_message":[],"_revision_number":13},{"id":"7ca46612eede7a83406047d2b7dfc5713f4644c8","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-15 14:02:56.000000000","message":"Patch Set 13:\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"682020d6e18519e6ef11414afc498a2bbd8a547f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-15 16:22:26.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c6fea9188c92cff64a05269410e341af72108c6a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-15 16:43:36.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"c64a7d02a421a07d0c047ecf18adc48db6830f44","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-15 16:54:22.000000000","message":"Patch Set 15:\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"cc684ebdee00627bfbd42294c155f2c179597473","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-05-15 17:16:28.000000000","message":"Patch Set 15: Code-Review+1\n\n(2 comments)\n\ncool beans!","accounts_in_message":[],"_revision_number":15},{"id":"bee2f43752c19d269d690828d3e96a4695b01908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 19:52:58.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a183df62bd94e208cce7fdb1202dbf0 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6036537205f465c8de876e6c1438631 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebdfefdae9644ee6825fba38b0b37d52 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5b5c08a807b4a8480ff95f7d034db84 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af4b8efca330433592d25c92b5eb092f : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5996cd03e9844973aa2c8872b9b03b4a : SUCCESS in 7m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/31bb81de7d90456e8f86df62c6b5e4a7 : SUCCESS in 6m 28s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/cc02b32497a74e578044a416d3078e2d : SUCCESS in 1h 48m 12s","accounts_in_message":[],"_revision_number":15},{"id":"ad66e1c016f3f5811119c65ea41cab4b2bc70961","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-05-18 13:19:19.000000000","message":"Patch Set 15: Code-Review+1\n\n(2 comments)\n\nWow Terry great job!","accounts_in_message":[],"_revision_number":15},{"id":"a07c102fabf504a35dd1b30ea6679d0865ea6bda","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2020-05-18 14:07:04.000000000","message":"Patch Set 15: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"5d88f7ed673d26fb1ec3c21a795100212111c911","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-18 14:48:55.000000000","message":"Patch Set 15:\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"931b9317c5691d50b8d2a3e6553cc331be8cce6d","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2020-05-19 15:29:28.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1\n\nLet\u0027s go! Thanks for considering the documentation as a follow-up!","accounts_in_message":[],"_revision_number":15},{"id":"c686aa0b3d8b8388c03a8b51df41c4ede6a771de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 15:29:39.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"5093c361055cd6e57f7e55ea50f0590823ccff57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 17:56:55.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e28d67f444c496a833a0322a8069f48 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade93c472cdd4de6aa6fd8ba1cd8205d : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd479266e2b1480c9954d4ac7f646d7a : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b42cc3aa13734eb492b31900066a2dc0 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/939547579c1244cb8ad6eb7c9fb07f68 : SUCCESS in 6m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/04a629484d504be5ac3d7651d245c6b2 : SUCCESS in 6m 26s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/bfa98ca52d3348a0a09f0fe63cc44665 : SUCCESS in 2h 24m 22s","accounts_in_message":[],"_revision_number":15},{"id":"3ad8831048aa2745d7f44fd3aaa6a724ca39891e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 17:56:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"c3fc2ea91e549bcb31f49c1b499ebd068110c8f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 17:59:04.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f3187e78e94bb7b76e09ca6d131985 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":15},{"id":"8941ca89e480311402e98efadaef09b3579858f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-30 19:51:18.000000000","message":"Created a revert of this change as I217d06b3ee2a04e7ab4995aef0696b791c228c6c","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"db0cdd9a4ffd64d99a6b012b5d0bf2c5bd4f96f4","revisions":{"e1c4e9efd74a5f28ba7cfa2efeaddf0e1f9270c7":{"kind":"REWORK","_number":1,"created":"2019-08-22 17:52:22.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/1"}}},"commit":{"parents":[{"commit":"68d462d355a9c9f7e6b12f08f4dfb0b0b3d443b2","subject":"Merge \"Add commands to set and get LRP options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/68d462d355a9c9f7e6b12f08f4dfb0b0b3d443b2"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-08-22 17:44:41.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-08-22 17:44:41.000000000","tz":-300},"subject":"WIP Add indexed column support to ovsdbapp","message":"WIP Add indexed column support to ovsdbapp\n\nThis currently creates an index for all tables with a single column\ndefined as an index in the schema, or have a lookup_table lookup\ndefined that is a single column on the table. This is a PoC code\ncheckin, so DNM.\n\nTODO:\n  make indexing opt-in to avoid changing behavior for users\n  make indexing easily configurable\n  general cleanup\n\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e1c4e9efd74a5f28ba7cfa2efeaddf0e1f9270c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e1c4e9efd74a5f28ba7cfa2efeaddf0e1f9270c7"}]},"branch":"refs/heads/master"},"36b84357aed808912bfc0ad9ec81dd06360d992b":{"kind":"REWORK","_number":2,"created":"2019-08-22 21:24:19.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/2"}}},"commit":{"parents":[{"commit":"68d462d355a9c9f7e6b12f08f4dfb0b0b3d443b2","subject":"Merge \"Add commands to set and get LRP options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/68d462d355a9c9f7e6b12f08f4dfb0b0b3d443b2"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-08-22 17:44:41.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-08-22 21:24:06.000000000","tz":-300},"subject":"WIP Add indexed column support to ovsdbapp","message":"WIP Add indexed column support to ovsdbapp\n\nThis currently creates an index for all tables with a single column\ndefined as an index in the schema, or have a lookup_table lookup\ndefined that is a single column on the table. This is a PoC code\ncheckin, so DNM.\n\nTODO:\n  make indexing opt-in to avoid changing behavior for users\n  make indexing easily configurable\n  general cleanup\n\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/36b84357aed808912bfc0ad9ec81dd06360d992b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/36b84357aed808912bfc0ad9ec81dd06360d992b"}]},"branch":"refs/heads/master"},"fec8794922e4da076460fdfe670f4e9775bb6350":{"kind":"REWORK","_number":3,"created":"2019-08-23 01:42:56.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/3"}}},"commit":{"parents":[{"commit":"68d462d355a9c9f7e6b12f08f4dfb0b0b3d443b2","subject":"Merge \"Add commands to set and get LRP options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/68d462d355a9c9f7e6b12f08f4dfb0b0b3d443b2"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-08-22 17:44:41.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-08-23 01:42:56.000000000","tz":-300},"subject":"WIP Add indexed column support to ovsdbapp","message":"WIP Add indexed column support to ovsdbapp\n\nThis currently creates an index for all tables with a single column\ndefined as an index in the schema, or have a lookup_table lookup\ndefined that is a single column on the table. This is a PoC code\ncheckin, so DNM.\n\nTODO:\n  make indexing opt-in to avoid changing behavior for users\n  make indexing easily configurable\n  general cleanup\n  add tests\n    ex: rows in lookup_table that we can\u0027t index, etc.\n\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fec8794922e4da076460fdfe670f4e9775bb6350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fec8794922e4da076460fdfe670f4e9775bb6350"}]},"branch":"refs/heads/master"},"686ab99bd81a8bfd5f3c92cccfb3794fc7c8c170":{"kind":"REWORK","_number":4,"created":"2019-08-23 03:57:29.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/4"}}},"commit":{"parents":[{"commit":"68d462d355a9c9f7e6b12f08f4dfb0b0b3d443b2","subject":"Merge \"Add commands to set and get LRP options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/68d462d355a9c9f7e6b12f08f4dfb0b0b3d443b2"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-08-22 17:44:41.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-08-23 03:56:42.000000000","tz":-300},"subject":"WIP Add indexed column support to ovsdbapp","message":"WIP Add indexed column support to ovsdbapp\n\nThis currently creates an index for all tables with a single column\ndefined as an index in the schema, or have a lookup_table lookup\ndefined that is a single column on the table. This is a PoC code\ncheckin, so DNM.\n\nTODO:\n  make indexing opt-in to avoid changing behavior for users\n  make indexing easily configurable\n  general cleanup\n  add tests\n    ex: rows in lookup_table that we can\u0027t index, etc.\n\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/686ab99bd81a8bfd5f3c92cccfb3794fc7c8c170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/686ab99bd81a8bfd5f3c92cccfb3794fc7c8c170"}]},"branch":"refs/heads/master"},"656c25b3260ba7cd551afedba6ff15083de78f30":{"kind":"REWORK","_number":5,"created":"2020-05-12 16:03:22.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/5"}}},"commit":{"parents":[{"commit":"2918d586c52b20e3360206ad53cac3c710c4104a","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2918d586c52b20e3360206ad53cac3c710c4104a"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-12 16:00:39.000000000","tz":0},"subject":"WIP Add indexed column support to ovsdbapp","message":"WIP Add indexed column support to ovsdbapp\n\nThis currently creates an index for all tables with a single column\ndefined as an index in the schema, or have a lookup_table lookup\ndefined that is a single column on the table. This is a PoC code\ncheckin, so DNM.\n\nTODO:\n  make indexing easily configurable\n  add tests\n    ex: rows in lookup_table that we can\u0027t index, etc.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/656c25b3260ba7cd551afedba6ff15083de78f30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/656c25b3260ba7cd551afedba6ff15083de78f30"}]},"branch":"refs/heads/master"},"ce371c2bcf97e7bf4f6682e887810697a689ebe8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-05-12 16:14:43.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/6"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-12 16:14:43.000000000","tz":0},"subject":"WIP Add indexed column support to ovsdbapp","message":"WIP Add indexed column support to ovsdbapp\n\nThis currently creates an index for all tables with a single column\ndefined as an index in the schema, or have a lookup_table lookup\ndefined that is a single column on the table. This is a PoC code\ncheckin, so DNM.\n\nTODO:\n  make indexing easily configurable\n  add tests\n    ex: rows in lookup_table that we can\u0027t index, etc.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ce371c2bcf97e7bf4f6682e887810697a689ebe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ce371c2bcf97e7bf4f6682e887810697a689ebe8"}]},"branch":"refs/heads/master"},"d9ec232436aa67a81873d37f0c64275ca6f3ad50":{"kind":"REWORK","_number":7,"created":"2020-05-12 21:25:15.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/7"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-12 16:47:37.000000000","tz":0},"subject":"WIP Add indexed column support to ovsdbapp","message":"WIP Add indexed column support to ovsdbapp\n\nThis currently creates an index for all tables with a single column\ndefined as an index in the schema, or have a lookup_table lookup\ndefined that is a single column on the table. This is a PoC code\ncheckin, so DNM.\n\nTODO:\n  make indexing easily configurable\n  add tests\n    ex: rows in lookup_table that we can\u0027t index, etc.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d9ec232436aa67a81873d37f0c64275ca6f3ad50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d9ec232436aa67a81873d37f0c64275ca6f3ad50"}]},"branch":"refs/heads/master"},"7db08d8ea1cacdc8db813cb9e7bd176e9472da31":{"kind":"REWORK","_number":8,"created":"2020-05-14 01:25:45.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/8"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-14 01:25:37.000000000","tz":0},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7db08d8ea1cacdc8db813cb9e7bd176e9472da31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7db08d8ea1cacdc8db813cb9e7bd176e9472da31"}]},"branch":"refs/heads/master"},"fe764d5c1d302395049c0db89a2369e03974f4c7":{"kind":"REWORK","_number":9,"created":"2020-05-14 01:48:06.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/9"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-14 01:48:02.000000000","tz":0},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fe764d5c1d302395049c0db89a2369e03974f4c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fe764d5c1d302395049c0db89a2369e03974f4c7"}]},"branch":"refs/heads/master"},"e45b187b912a6631c180a2da9fc9ed36f2f19827":{"kind":"REWORK","_number":10,"created":"2020-05-14 19:35:24.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/10"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-14 16:46:22.000000000","tz":0},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e45b187b912a6631c180a2da9fc9ed36f2f19827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e45b187b912a6631c180a2da9fc9ed36f2f19827"}]},"branch":"refs/heads/master"},"c18a2264caa556bbe44bb211f2fe3cca72e8326d":{"kind":"REWORK","_number":11,"created":"2020-05-15 00:01:54.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/11"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-15 00:01:53.000000000","tz":0},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c18a2264caa556bbe44bb211f2fe3cca72e8326d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c18a2264caa556bbe44bb211f2fe3cca72e8326d"}]},"branch":"refs/heads/master"},"8456e7f99137f349d9715fca33a9579ff26eea92":{"kind":"REWORK","_number":12,"created":"2020-05-15 03:52:11.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/12"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-15 03:52:07.000000000","tz":0},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8456e7f99137f349d9715fca33a9579ff26eea92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8456e7f99137f349d9715fca33a9579ff26eea92"}]},"branch":"refs/heads/master"},"91b469911c6a76c86c8293f66272a0c199738d88":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2020-05-15 04:18:56.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/13"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-15 04:16:47.000000000","tz":0},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nThere are slight API changes w/ Backend.ovsdb_connection\nproperties added and start_connection becoming an instance method\ninstead of a classmethod. But since ovsdb_connection was the\nitem being stored on the class and it still is via an overridable\nproperty, users shouldn\u0027t be broken by the change.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/91b469911c6a76c86c8293f66272a0c199738d88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/91b469911c6a76c86c8293f66272a0c199738d88"}]},"branch":"refs/heads/master"},"5305a68c3293a3096505124b1e475e897f4c5949":{"kind":"REWORK","_number":14,"created":"2020-05-15 16:22:26.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/14"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-15 16:22:17.000000000","tz":0},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nThere are slight API changes w/ Backend.ovsdb_connection\nproperties added and start_connection becoming an instance method\ninstead of a classmethod. But since ovsdb_connection was the\nitem being stored on the class and it still is via an overridable\nproperty, users shouldn\u0027t be broken by the change.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5305a68c3293a3096505124b1e475e897f4c5949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5305a68c3293a3096505124b1e475e897f4c5949"}]},"branch":"refs/heads/master"},"db0cdd9a4ffd64d99a6b012b5d0bf2c5bd4f96f4":{"kind":"REWORK","_number":15,"created":"2020-05-15 16:43:36.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/76/678076/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/76/678076/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/76/678076/15"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-15 16:43:32.000000000","tz":0},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nThere are slight API changes w/ Backend.ovsdb_connection\nproperties added and start_connection becoming an instance method\ninstead of a classmethod. But since ovsdb_connection was the\nitem being stored on the class and it still is via an overridable\nproperty, users shouldn\u0027t be broken by the change.\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/db0cdd9a4ffd64d99a6b012b5d0bf2c5bd4f96f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/db0cdd9a4ffd64d99a6b012b5d0bf2c5bd4f96f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
