)]}'
{"id":"openstack%2Fovsdbapp~680297","triplet_id":"openstack%2Fovsdbapp~master~Ia1a7030878f12ca0b8184363173eada929ece3bd","project":"openstack/ovsdbapp","branch":"master","topic":"build-pdf-docs","hashtags":[],"change_id":"Ia1a7030878f12ca0b8184363173eada929ece3bd","subject":"Update Zuul jobs: Enable docs publishing","status":"MERGED","created":"2019-09-05 06:06:48.000000000","updated":"2019-09-26 12:04:06.000000000","submitted":"2019-09-26 11:59:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680297-1569499176211-e985b01a","meta_rev_id":"c7d423b3acabe03c68211e4809456f9dd717b9cb","_number":680297,"virtual_id_number":680297,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-10 19:44:33.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-26 11:59:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"recommended":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-09-06 13:14:55.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-09-10 17:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2019-09-12 07:43:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2019-09-10 17:14:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 13:14:55.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-09-10 17:14:17.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-09-12 07:43:30.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2019-09-26 11:59:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6603144c3125995b6eed1cd4a8061204e6b8238a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-05 06:06:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3c1c90c46fcd5a96e3a60350ce95fec6cc85c7b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-05 06:07:27.000000000","message":"Topic changed from update-jobs to build-pdf-docs","accounts_in_message":[],"_revision_number":1},{"id":"b6e804fa1b16de85080956225d47df5ab4561aca","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-05 06:18:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7542033510879090a68f8b10ee174401c72207ea","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-05 06:18:42.000000000","message":"Topic changed from update-jobs to build-pdf-docs","accounts_in_message":[],"_revision_number":2},{"id":"fdedf9700887c0c97d8f82e8d4a7c27bb9cce661","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-05 07:17:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a0aa6bbfc2042138651774b0bf1924be6339bce7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 08:00:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a6706aebae984c4b90f5143202582bc3 : SUCCESS in 2m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2ab092a5f3d4317966c5ee8788fcac8 : SUCCESS in 2m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/671c6860b5694a369ba4dbfd8bbb7a8d : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d45de8a578ee4d72aba151929b76b487 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9005da673b294bd28fe185a71e046ae6 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/284c4c3442ed42eabc8f9dfc32444a60 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd764cc7e7104371bf9039da56427909 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb2ce58edec45eea87ba33ddfb92da7 : SUCCESS in 5m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f0ccbebe50bd4ab79d60af2c26e3b66a : SUCCESS in 4m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2db94641722b49f8ac05fd862b6eb7b9 : SUCCESS in 5m 19s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/2789af4f5d4846818ce50dd9bfc7bd53 : FAILURE in 22m 19s","accounts_in_message":[],"_revision_number":3},{"id":"02ac6de21a1dea75027df5b25c8cd1aa0ad1522f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-05 10:25:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2927a1d54cbc609b6d71feca43df04510d30d620","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 11:04:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9bb00f446f5f4b7bbc434e717fd3711a : SUCCESS in 2m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f41d4d343c9740eea50262398933d554 : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8432261317344a4185ee68238f915175 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a5d6d03a5a24db59e4323610aea90df : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16b1e94098774d9eab39de385a8025d3 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9f1a7298d1d4990a4184cb392bf1bd0 : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e539bf3472924ad5990715b7f5143d3e : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f812b72efa2a45dc877cb757a5891f82 : SUCCESS in 3m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/48a35df5484f401c891b1556a133438a : SUCCESS in 5m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/730e453273444369a7fcfbd0c5b8d9b8 : SUCCESS in 5m 51s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/318419f50b414bb29bb0c64914b5ab57 : FAILURE in 22m 58s","accounts_in_message":[],"_revision_number":3},{"id":"67b0e50bb65442586829c85bf0027110a2a5d625","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-06 13:14:43.000000000","message":"Patch Set 3:\n\nrecheck the tempest networking-ovn job was fixed","accounts_in_message":[],"_revision_number":3},{"id":"9cc5bf8a35408241e0e3bbd089c4d68ee4e2f68d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-06 13:14:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f2284e4e14a053d2c2f4c525ca1da90d086086a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:38:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7a2abd8d13e849ee9b0066288293a466 : SUCCESS in 2m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7ddb2f30cf544b6e9688550f276026a5 : SUCCESS in 2m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd1565a7df5e4495a6545d64f9a1652d : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bd4b2e6258642b4bc9ebc251f4283a3 : SUCCESS in 2m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab9ae7b1eccf4d19a7f6e76cd9c574a5 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cc306d045c744e2b493c46abe6c44d2 : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ca811b9ecdf24334b93240e202cf946f : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6281b7787a24df9936cfaa80990eb62 : SUCCESS in 2m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6bb3a159d4364ff6b3b7af1838b2da63 : SUCCESS in 5m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5514b290f7614041845359ce02bab2cd : SUCCESS in 4m 46s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/1fc8cf485dfb4b138829da79015aba23 : SUCCESS in 2h 16m 37s","accounts_in_message":[],"_revision_number":3},{"id":"88316208a2e4ca8196d4126e5c4c2d5fc0dd4165","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-10 17:14:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"23d597fa14c0ff322c34d463052154ef44719d39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 17:15:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f0f4faaee20bd85a3f8aef4800bc51e833368243","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 19:40:44.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/be7f22cc15e74920a64b4a256199c326 : SUCCESS in 3m 52s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8951fbe219a34e49b3ac3ce5828e0120 : SUCCESS in 2m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47c11da4b45c40648f2e6d0cbb2ea353 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dfddb6eac8b4eb8b2ed8cec13573598 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a4bdf3635c346489efe35d16f9f5fd6 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e60f01fff1f4121b0543fe790f9589d : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/980d854538154fa3abe5e154a5d08fe8 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd1b54fa9fb41378ba4b889035c3948 : SUCCESS in 3m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/94e5cbd2f608408a8c5b91a7779c9b25 : FAILURE in 5m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad604bbfd07844468c025c3efa6da065 : FAILURE in 6m 12s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/b814adc20e7a4b96b07d34087ab163be : SUCCESS in 2h 19m 23s","accounts_in_message":[],"_revision_number":3},{"id":"b9ad33dc985e00baf91ab98e86f861d118b4186a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-10 19:44:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c5878e6fed329d4eda5dfd365792a1f1bf210978","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 22:12:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ea39b8bb92d84a6e8af64501a2377890 : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4ad22a87af364426b32cff1c4e3a2560 : SUCCESS in 2m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6836b60b1994470f8414917d4077535a : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c915331b21b549afbab8d4bd11ecdc53 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/022704541cbb4baf97b3750ab195f9de : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0f5e75d124c437db721b5a9e1289ff5 : SUCCESS in 3m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0d80560c1e649b99cfd969e08912dd2 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0166fdaf524a9ab7934c75e092eb57 : SUCCESS in 3m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1f4e5c18b5ff47fba1ac26d8bb62f7e1 : FAILURE in 5m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/276124dcb38440db832207918daf346b : FAILURE in 4m 36s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/c1fbf53f301046ababc8a8fdb2d1c82b : SUCCESS in 2h 08m 18s","accounts_in_message":[],"_revision_number":3},{"id":"be9982b8026f601f18a503aff7d6a25d7b3a74ff","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2019-09-12 07:43:30.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"116a5c0eac5a0e9c798b0d69018db557f4506c6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 10:31:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6cbf926f7e7f4c11863a05ece625b76c : SUCCESS in 2m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/832f4685ee6e47dbaa83b7665b5b9437 : SUCCESS in 2m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f859bde527e04402a965603fa3d470c1 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f021a746b5964f5f833653f5f81299fe : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bbebd5d5e0b642c1bc1399c1073d736a : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4ec6a1a88a0462d91a7b1c8ddc3ef71 : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0954a50fd16f4623afb2d5c2520621d4 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e394e8d16874e69b7dd74789b7bfc8c : SUCCESS in 3m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/25b9be545bfe4c108c59b7fe1b110e96 : FAILURE in 4m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8acf2e6ab98642749c8712d01bafd36d : FAILURE in 4m 29s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/87db18041d2842d3beb0c33dabb4ffe4 : SUCCESS in 2h 21m 01s","accounts_in_message":[],"_revision_number":3},{"id":"885ac4ae9441ba3274fbd7002db890596076202e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-09-26 06:26:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4d75cba93be8af9df7f66845abd541b52e2a84b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:11:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0b3b2b36732f45b787e71c84b6749ba8 : SUCCESS in 4m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4fbf5ba3fdf242cb8efc7e91ddc27b9a : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9b61112d8b349ba8ccecc1e6069429d : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2627519da9f4ff29019190dfe218361 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b23be9bb565a49f68428da7fc1e323ac : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46c42b40b4cb403983c5537887bdb72a : SUCCESS in 6m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd5ec95ad20149f1bfad1662ede7f6b3 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cf0e3d83a1b44ca9f4a27579f656895 : SUCCESS in 8m 50s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1ea41a013adb4536bc4f14a7da9479cf : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ef02214690e46bbaf9a2b7fcaf6e5b2 : SUCCESS in 8m 19s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/6a57f914dfea44329b619e0cbdee1b2e : SUCCESS in 2h 21m 08s","accounts_in_message":[],"_revision_number":3},{"id":"aaf547f4e1be8f44d1eae64dfc5ce4c7aecd3ae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 09:11:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d13538966f0abcb330f3c7f9577a6d6b660a6fa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 11:59:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6cecf0d45175db225c1a02ebcdd96c491b085aa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 11:59:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/31333b20b47a4277bf06c669ae09727d : SUCCESS in 2m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5710a6c2e1724ed9b4a2a7a274824067 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b022dd125b7c40b093efa9384e1eed48 : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8aaf59ae54b4237a21aeac9c97737ee : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d103c7c0848c496ebddf75cc62c7820e : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d74f0ae5a2514cc992706365db58b1c0 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acd6df19776c436cb5772bd423fbb639 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728c88cba30b420ca871339249e525cc : SUCCESS in 4m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9f7514496370443481840fcde4de21d4 : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd58e72c40d04869a7dbe308407432cf : SUCCESS in 5m 49s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/02e4e8ba0a9b443ba24f4a0e296ab92b : SUCCESS in 2h 45m 18s","accounts_in_message":[],"_revision_number":3},{"id":"c7d423b3acabe03c68211e4809456f9dd717b9cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-26 12:04:06.000000000","message":"Patch Set 3:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21416c2e2a164375b1cca792ac5cfcd0 : FAILURE in 2m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ac3d3de9ecc3bd26c92f3480c067667771010d7e","revisions":{"51e430b52c55eb0f1776b1f555580d024557eda6":{"kind":"REWORK","_number":1,"created":"2019-09-05 06:06:48.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/97/680297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/97/680297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/1"}}},"commit":{"parents":[{"commit":"74b52bb3491a9c4f6ec40e15d74652fc415269e0","subject":"Add HA Chassis Group related commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/74b52bb3491a9c4f6ec40e15d74652fc415269e0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-05 06:01:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-05 06:01:07.000000000","tz":120},"subject":"Update Zuul jobs: Enable docs publishing","message":"Update Zuul jobs: Enable docs publishing\n\nStart publishing docs with using the proper template.\n\nAlso, use lower-constraints template instead of individual jobs.\n\nChange-Id: Ia1a7030878f12ca0b8184363173eada929ece3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/51e430b52c55eb0f1776b1f555580d024557eda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/51e430b52c55eb0f1776b1f555580d024557eda6"}]},"branch":"refs/heads/master"},"ffba596846e0bef98252e23d16977a3e6c9085ef":{"kind":"REWORK","_number":2,"created":"2019-09-05 06:18:05.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/97/680297/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/97/680297/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/2"}}},"commit":{"parents":[{"commit":"74b52bb3491a9c4f6ec40e15d74652fc415269e0","subject":"Add HA Chassis Group related commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/74b52bb3491a9c4f6ec40e15d74652fc415269e0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-05 06:01:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-05 06:16:23.000000000","tz":120},"subject":"Update Zuul jobs: Enable docs publishing","message":"Update Zuul jobs: Enable docs publishing\n\nStart publishing docs with using the proper template.\n\nFor this, add also a doc/requirements.txt file and move docs\nrequirements there. Update tox.ini for this and use the modern way to\ninvoke sphinx-build for building.\nRemove obsolete sectoins from setup.cfg.\nUpdate the conf.py files and remove unused lines\nwith newer openstackdocstheme - and require newer theme.\n\nAlso, use lower-constraints template instead of individual jobs.\n\nChange-Id: Ia1a7030878f12ca0b8184363173eada929ece3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ffba596846e0bef98252e23d16977a3e6c9085ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ffba596846e0bef98252e23d16977a3e6c9085ef"}]},"branch":"refs/heads/master"},"ac3d3de9ecc3bd26c92f3480c067667771010d7e":{"kind":"REWORK","_number":3,"created":"2019-09-05 07:17:43.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/97/680297/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/97/680297/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/97/680297/3"}}},"commit":{"parents":[{"commit":"74b52bb3491a9c4f6ec40e15d74652fc415269e0","subject":"Add HA Chassis Group related commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/74b52bb3491a9c4f6ec40e15d74652fc415269e0"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-05 06:01:07.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-09-05 07:17:05.000000000","tz":120},"subject":"Update Zuul jobs: Enable docs publishing","message":"Update Zuul jobs: Enable docs publishing\n\nStart publishing docs with using the proper template.\n\nFor this, add also a doc/requirements.txt file and move docs\nrequirements there. Update tox.ini for this and use the modern way to\ninvoke sphinx-build for building.\nRemove obsolete sectoins from setup.cfg.\nUpdate the conf.py files and remove unused lines\nwith newer openstackdocstheme - and require newer theme.\n\nFix indents in CONTRIBUTING.rst so that the formatting works nicely.\n\nAlso, use lower-constraints template instead of individual jobs.\n\nChange-Id: Ia1a7030878f12ca0b8184363173eada929ece3bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ac3d3de9ecc3bd26c92f3480c067667771010d7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/ac3d3de9ecc3bd26c92f3480c067667771010d7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
