)]}'
{"id":"openstack%2Fovsdbapp~683688","triplet_id":"openstack%2Fovsdbapp~master~I264200ae016e3928fb4aff3bdcc7c1a94f91cae4","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"I264200ae016e3928fb4aff3bdcc7c1a94f91cae4","subject":"Handle the ovs/ovn split","status":"MERGED","created":"2019-09-20 18:24:06.000000000","updated":"2019-10-02 18:33:25.000000000","submitted":"2019-09-24 21:02:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683688-1569358966743-844a795a","meta_rev_id":"0f5ff66778397970cde7656fc8a36dc56af4e382","_number":683688,"virtual_id_number":683688,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 21:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":2,"date":"2019-09-24 18:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-24 11:57:21.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":1,"date":"2019-09-23 13:34:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 11:57:21.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-09-24 21:02:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9dfa37142124260686da11f2a46bd9d68f7891c1","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-20 18:24:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"259df12683bb64236f28afaf01e62d3259dff3e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 20:30:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54926a7e09fa4164bb3215bf471c2606 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21753d51acf241e1bd9ecf98d7d57747 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a624ab10450c45499150a75f916ca2e8 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12dd9a2490bb47b8af409cc28a06f093 : SUCCESS in 3m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b61915155bf4a52b64a4c10ea18862a : SUCCESS in 4m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/be5549e1bd054ca9ab9a62da919cde84 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/53f781b524344c16899c1b9a537e965c : SUCCESS in 4m 46s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/c6552264f53d40c3a3b12e7b45cca88d : SUCCESS in 1h 58m 50s","accounts_in_message":[],"_revision_number":1},{"id":"55cd5a428eba506d63fb61d901a7bab42fa9a630","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-23 08:10:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3562b52c95d55146f02be27b0ba8cd695f089153","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-23 13:34:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e641a6d1d5cc288d0bdef3d00ff61439c9be9c7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 13:34:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6cb4a03634d6587cf3e2d807f39067563d70d28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-23 16:35:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0f7c3e219614f69a49cb7acb48d594f : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59de47d75c5c4076b7783d5acd6c6078 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85859199f41e48509710ac558a7da66b : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e15899e3a33f479eb311eb57752eee14 : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c87e379f95344d39e1c34cd92997578 : SUCCESS in 3m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d18c50283e38495fa0fb0b07feb12d3c : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b65bc3b4d4c04fac8627e261e111fbea : SUCCESS in 5m 57s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/1f12457cad144fce93061a47c3390599 : FAILURE in 2h 57m 46s","accounts_in_message":[],"_revision_number":1},{"id":"4d30793ce107372cb4c9d47886ac84bfbf97fcf0","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-24 01:26:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fb28901aa6454dd781f000c93f82af7dbde52c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 03:47:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153b5221e57a48eba454aa31b0f027e2 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac6a92aeab494c7390821460b884f6a9 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15751114e9604e7f823db74a16d37a79 : SUCCESS in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09a8eddb132d48a798950b509ef47649 : SUCCESS in 3m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/603b8fd6a6e04b22927d0f0ab7dcd20f : SUCCESS in 4m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/16ea4106a04a4a36b89045b0a66ed3e0 : FAILURE in 4m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ac7ec6b865946ec937a7d22c5259d5b : FAILURE in 5m 41s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/8177fcba7b55486ba73e59bd64714c25 : SUCCESS in 2h 18m 35s","accounts_in_message":[],"_revision_number":1},{"id":"e8d3f80e4a79c40435da5bd784091acfa49db6ed","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-09-24 11:57:21.000000000","message":"Patch Set 1: -Code-Review\n\nNope, seems like it broke functional tests","accounts_in_message":[],"_revision_number":1},{"id":"2e867471a03adc66850555e9b73428546e693a33","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-24 12:24:24.000000000","message":"Patch Set 1:\n\nrecheck\n\nI don\u0027t think this broke it, I OVN stopped building for a bit since the last commit there is \"fix compilation errors\".","accounts_in_message":[],"_revision_number":1},{"id":"0ca0fcb9c76242d703504e5044b6713b7b6c1aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:17:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24ea8dc3252b48128463be1010265a36 : SUCCESS in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12cb84ac391449c6beb519847a62356e : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75a4921d77e74f98b8b335eb9a240521 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f948183920274fe5b609592b952092bc : SUCCESS in 3m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/480940a9f4db449b9f49b0f6e899866c : SUCCESS in 4m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/abdf355bb9e944e38ae81b36c33a7d77 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/369998cb17ea4882af4d8a32bcd39215 : SUCCESS in 5m 02s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/07e178aa3bca415dbf9c413a4de4dca2 : SUCCESS in 2h 48m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e729c33d81e5a767083df8478b7535337a15dfce","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-09-24 18:27:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"89b519f4076c465a579b503737813c340e1dfbe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 18:28:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e6da9a634df7fc7fbeb2634156c2326a0529064","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 21:02:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"768c88e599db60d1f7121d7717e5175c6a1054cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 21:02:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ff42d74857a47688cc0ae2b5b8aba97 : SUCCESS in 12m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd6fa14dc7d34e3c9883ae5d8bafdaa6 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a4689ea44334b95869fe445ed819d24 : SUCCESS in 11m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f22a859e64fb4ae3a9ab2c784ceb5402 : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cac44c6a7c6496897cc79fcaff912b1 : SUCCESS in 11m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1cf6559cc0e545ee9a4ddac566c87dc7 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8ac3ebac11f4444ebcd409090bf00a09 : SUCCESS in 5m 28s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/4c7bbc279937499c83f7bf133a2452b9 : SUCCESS in 2h 25m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0f5ff66778397970cde7656fc8a36dc56af4e382","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-02 18:33:25.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit fa5744b82b5b11c01c0c6d186604b206f0237624","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54ee85113a041d7abd16b10c65d70078a1e76bff","revisions":{"54ee85113a041d7abd16b10c65d70078a1e76bff":{"kind":"REWORK","_number":1,"created":"2019-09-20 18:24:06.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/88/683688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/88/683688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/88/683688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/88/683688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/88/683688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/88/683688/1"}}},"commit":{"parents":[{"commit":"51f675ac4b51bd9a82579bb2e97de7783b689bae","subject":"Merge \"Fix the url, change http to https\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/51f675ac4b51bd9a82579bb2e97de7783b689bae"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-20 18:22:35.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-20 18:24:05.000000000","tz":-300},"subject":"Handle the ovs/ovn split","message":"Handle the ovs/ovn split\n\nThis is the bare minimum to run against ovs/ovn master after the\nsplit. It should be updated to handle running against system-\ninstalled OVN.\n\nChange-Id: I264200ae016e3928fb4aff3bdcc7c1a94f91cae4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/54ee85113a041d7abd16b10c65d70078a1e76bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/54ee85113a041d7abd16b10c65d70078a1e76bff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
