)]}'
{"id":"openstack%2Fovsdbapp~686238","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~I264200ae016e3928fb4aff3bdcc7c1a94f91cae4","project":"openstack/ovsdbapp","branch":"stable/train","topic":"686238","hashtags":[],"change_id":"I264200ae016e3928fb4aff3bdcc7c1a94f91cae4","subject":"Fix all of the train gate issues","status":"MERGED","created":"2019-10-02 18:33:25.000000000","updated":"2019-10-04 18:54:44.000000000","submitted":"2019-10-04 17:04:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686238-1570208653642-b42f53c0","meta_rev_id":"a1e00da7a431e9db09bb2ee6f6c984b86e332887","_number":686238,"virtual_id_number":686238,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 17:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-04 16:37:20.000000000","_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2019-10-04 16:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2019-10-04 16:48:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 01:55:09.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2019-10-04 12:45:16.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-10-04 13:21:32.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-10-04 16:48:26.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-10-04 17:04:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92c8709db506de15df0f221fcaea156f2d09e089","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-02 18:33:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a176aaba58c80cf32b1c5d0bd0ce89e82d95f830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 18:45:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc3f7530a4d4fdcb2bbb65797d49d88 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6274c90b1b440fe87ee1ac052dfeb02 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1297c2c1b4134a15b0665143cbb4b5d6 : SUCCESS in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d18427ac1b44ea690d8b019a9275cf1 : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f89b3b5f0da3490eab2e6be1e23adb32 : SUCCESS in 4m 18s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7d1f9832087948bb9d1c17bc58112e3b : FAILURE in 3m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4cabb20f843475ba59bd94d2af1cf06 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ff967c4c630b27e0e0f37349045fd87c42c1eef7","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-03 20:47:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3939913c7434a2f7aef14e6b520fdde8d6f9482b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 20:59:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/656ff97808444399973918e8fb338ab0 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c88b6a57cca42c0bfb167bec548217c : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3b64e070b1840aea749de70c750621a : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7335dd8ff4f46d4b39345c7409c60fa : SUCCESS in 3m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c59f30a1c9de4b45acc112425e964734 : SUCCESS in 3m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/312f71cbd341479eb612dc341fc477c6 : FAILURE in 3m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f45c8b55aafb4e4c81f5b0c08cc613e4 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"da9c6dcf2fbf6561a3472ff2ba8b275ecee90481","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-03 22:21:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1b45e449ed9b339ba73b3a1c4c21c6723e70c962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 22:37:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e5a67540304c61b9a90945cd575d3c : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3df8271229464b3e95662adff58c49b4 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/599d313bb7ab47d58a13e649b4422941 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0879837d7cdc4273843d0ed71e5e665c : SUCCESS in 4m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47d0795f9b61403298d717f07f9cb7ac : SUCCESS in 4m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bd8d83fd84f946ee8da65a416c6840c9 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"92414baf7cf6ecbba04b396d96d94704b15be998","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-04 01:55:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bee0d79150b7e75dde226a0fcdd58b36574b1734","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2019-10-04 12:45:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a336907cfdc27046206bdad43988a12515ba4e82","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 13:05:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"05cc15df27ab0ca053772fec0897008abfc83a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 13:18:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b4ef522ebb47ecb30dab86175ee7d5 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5401c457d9414d398b7d1ce0a7e411eb : SUCCESS in 2m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/816cc540400c40a4a5797cbb470b7411 : SUCCESS in 8m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94b0db5f06f141b89bf6d797e3acf082 : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/334539bd124e45aaa735e4feeda1a950 : SUCCESS in 6m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fbf6aafc135a4f9b862e30afe58596f8 : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":3},{"id":"f28ec4abeb669a7f0d89d7e65c1f32eb6f36c406","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-10-04 13:21:32.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM, I can only +1 here","accounts_in_message":[],"_revision_number":3},{"id":"2df43497bacb513e8bc7fc3df2374f4769143109","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 13:30:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"eb659712d34390d975de89145895936442d13974","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:27:19.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"efb3246da0aa5e98a52fd105c3f3d440f3bc6ad8","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:32:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e39056844745c7aaffa62c8fdd980ddf4e999cf1","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:33:35.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e30b84df56fc155dfcc22cd2335d0d2106892818","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-04 16:36:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3f66a2ab09c3af31106546a69bc887cf2ceb0f76","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:37:20.000000000","message":"Patch Set 5:\n\nyes, but it didn\u0027t merge because of the other gate issues.","accounts_in_message":[],"_revision_number":5},{"id":"d9eb5e909fa45def62956a4ee845cbea77048434","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-04 16:37:58.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"36ff59624205269578838874edb942ce01e3b9b6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:42:45.000000000","message":"Abandoned\n\npart of squash here https://review.opendev.org/#/c/686238/","accounts_in_message":[],"_revision_number":5},{"id":"af0939f505f6892de68fcc162ab09467acc85109","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:43:30.000000000","message":"Restored\n\nI swear...","accounts_in_message":[],"_revision_number":5},{"id":"feb130e1500c797413417465b6d5b2e31fff161d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-04 16:48:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nninjamerging to fix gate issues","accounts_in_message":[],"_revision_number":5},{"id":"caa5e01e81ef4569e539ae6a3a92cedae938d8a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 16:53:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21624d66d4ee4409b9fb2f0f56e20b61 : SUCCESS in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cccdb8290dfa46128bfbf91aecb8271b : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8488820d90894c57a7f8687e52e68d29 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac541c8f132549eaa73cc601a176c364 : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa12326eeb2c4bb9844c31f01770e4da : SUCCESS in 4m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/14d7985d4e6349ec8619237162427bfd : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":5},{"id":"69d918189008551135abc75dda9181976afbefac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 16:53:25.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4c8be6ef1d351cdb460ae0da1a6de2dfb5cc281d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 17:04:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d733851d8caa4787bdd7d70c367435be : SUCCESS in 3m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/96e9aad5143045d9944fcab19f111482 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4d3db98f32b44c4bedf1cd2ac80127d : SUCCESS in 2m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/82922232a58e4e9ca7a6b79162d5aec4 : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe6d84f532f74852a8a35c2487f9bfce : SUCCESS in 3m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1ed093680b9b42bf9feda8658f045bdf : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":5},{"id":"d2218d95eb0e40f7939ab85c5eb0b84cf5e8c273","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 17:04:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"c5a393e14932db8fb4928f93334d0a8744de3edd","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2019-10-04 18:54:44.000000000","message":"Patch Set 5:\n\nnice!","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"498e0863f42e845ecd02e33c8e1a8c47c22ec017","revisions":{"fa5744b82b5b11c01c0c6d186604b206f0237624":{"kind":"REWORK","_number":1,"created":"2019-10-02 18:33:25.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/38/686238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/38/686238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/1"}}},"commit":{"parents":[{"commit":"74b52bb3491a9c4f6ec40e15d74652fc415269e0","subject":"Add HA Chassis Group related commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/74b52bb3491a9c4f6ec40e15d74652fc415269e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-20 18:22:35.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-02 18:33:25.000000000","tz":0},"subject":"Handle the ovs/ovn split","message":"Handle the ovs/ovn split\n\nThis is the bare minimum to run against ovs/ovn master after the\nsplit. It should be updated to handle running against system-\ninstalled OVN.\n\nChange-Id: I264200ae016e3928fb4aff3bdcc7c1a94f91cae4\n(cherry picked from commit 54ee85113a041d7abd16b10c65d70078a1e76bff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fa5744b82b5b11c01c0c6d186604b206f0237624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fa5744b82b5b11c01c0c6d186604b206f0237624"}]},"branch":"refs/heads/stable/train"},"eccd988db846df1227f25765a0d598b91f7d05c8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-10-03 22:21:02.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/38/686238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/38/686238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/2"}}},"commit":{"parents":[{"commit":"c9ffbfde7ee639ebe59175ab65728f56d1f3b099","subject":"Fix building OVS after python2 removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c9ffbfde7ee639ebe59175ab65728f56d1f3b099"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-20 18:22:35.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-03 22:21:02.000000000","tz":0},"subject":"Handle the ovs/ovn split","message":"Handle the ovs/ovn split\n\nThis is the bare minimum to run against ovs/ovn master after the\nsplit. It should be updated to handle running against system-\ninstalled OVN.\n\nChange-Id: I264200ae016e3928fb4aff3bdcc7c1a94f91cae4\n(cherry picked from commit 54ee85113a041d7abd16b10c65d70078a1e76bff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/eccd988db846df1227f25765a0d598b91f7d05c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/eccd988db846df1227f25765a0d598b91f7d05c8"}]},"branch":"refs/heads/stable/train"},"3c4d1b8f0c9de87803b341745ea74d9e919acb92":{"kind":"NO_CHANGE","_number":3,"created":"2019-10-04 13:05:01.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/38/686238/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/38/686238/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/3"}}},"commit":{"parents":[{"commit":"bf4ec880f5a613c76557235bbd0331bdcbeaf760","subject":"Fix building OVS after python2 removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bf4ec880f5a613c76557235bbd0331bdcbeaf760"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-20 18:22:35.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 13:05:01.000000000","tz":0},"subject":"Handle the ovs/ovn split","message":"Handle the ovs/ovn split\n\nThis is the bare minimum to run against ovs/ovn master after the\nsplit. It should be updated to handle running against system-\ninstalled OVN.\n\nChange-Id: I264200ae016e3928fb4aff3bdcc7c1a94f91cae4\n(cherry picked from commit 54ee85113a041d7abd16b10c65d70078a1e76bff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3c4d1b8f0c9de87803b341745ea74d9e919acb92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/3c4d1b8f0c9de87803b341745ea74d9e919acb92"}]},"branch":"refs/heads/stable/train"},"58a69d28c081135ff35fafb788304598678c2dfb":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-04 16:32:05.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/38/686238/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/38/686238/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/4"}}},"commit":{"parents":[{"commit":"7574897bc0474aef850c9822c582cebeab882da5","subject":"Fix building OVS after python2 removal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7574897bc0474aef850c9822c582cebeab882da5"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-09-20 18:22:35.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 16:32:05.000000000","tz":0},"subject":"Handle the ovs/ovn split","message":"Handle the ovs/ovn split\n\nThis is the bare minimum to run against ovs/ovn master after the\nsplit. It should be updated to handle running against system-\ninstalled OVN.\n\nChange-Id: I264200ae016e3928fb4aff3bdcc7c1a94f91cae4\n(cherry picked from commit 54ee85113a041d7abd16b10c65d70078a1e76bff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/58a69d28c081135ff35fafb788304598678c2dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/58a69d28c081135ff35fafb788304598678c2dfb"}]},"branch":"refs/heads/stable/train"},"498e0863f42e845ecd02e33c8e1a8c47c22ec017":{"kind":"REWORK","_number":5,"created":"2019-10-04 16:33:35.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/38/686238/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/38/686238/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/38/686238/5"}}},"commit":{"parents":[{"commit":"74b52bb3491a9c4f6ec40e15d74652fc415269e0","subject":"Add HA Chassis Group related commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/74b52bb3491a9c4f6ec40e15d74652fc415269e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 16:19:39.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 16:32:48.000000000","tz":-300},"subject":"Fix all of the train gate issues","message":"Fix all of the train gate issues\n\nUpdate gitreview for train\n\nChange-Id: Ieab25244e3d1bff56537de1d5915736f3e8c26dc\n\nFix building OVS after python2 removal\n\nSwitch functional to use python3 to install python3-based\ndependencies, which the OVS build checks for. Also, remove\nthe py36 test\n\nChange-Id: I2d9599d7050a15f459e6845a112ffc9948b2a7bd\n(cherry-picked from commit d5420ab5d248b93c5ce1c65d29521dc52ad3c5ae)\n\nHandle the ovs/ovn split\n\nThis is the bare minimum to run against ovs/ovn master after the\nsplit. It should be updated to handle running against system-\ninstalled OVN.\n\nChange-Id: I264200ae016e3928fb4aff3bdcc7c1a94f91cae4\n(cherry picked from commit 54ee85113a041d7abd16b10c65d70078a1e76bff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/498e0863f42e845ecd02e33c8e1a8c47c22ec017"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/498e0863f42e845ecd02e33c8e1a8c47c22ec017"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
