)]}'
{"id":"openstack%2Fovsdbapp~686416","triplet_id":"openstack%2Fovsdbapp~master~I30125ee023dfa59c4a5dde37195109e379e55c4b","project":"openstack/ovsdbapp","branch":"master","topic":"nested-transaction","hashtags":[],"change_id":"I30125ee023dfa59c4a5dde37195109e379e55c4b","subject":"Change EAFP to LBYL style in nested transaction","status":"MERGED","created":"2019-10-03 14:35:36.000000000","updated":"2019-10-08 18:53:20.000000000","submitted":"2019-10-08 18:50:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686416-1570560649343-8e81ebfc","meta_rev_id":"1168b431225098a332ae6076fe39aaf26076713e","_number":686416,"virtual_id_number":686416,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-08 18:50:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-04 12:16:20.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":2,"date":"2019-10-04 17:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":1,"date":"2019-10-08 16:39:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-08 16:39:07.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-10-08 18:50:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71ffe0ff82224a3aa64378cbbcf3b71a642563f0","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-03 14:35:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af40899db346a2d27800be1110d2d1a321141605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 17:12:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5f1aa39f4c643acb72c1cd8dfae5c5b : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78cb438eae2942fcb6edbc95cebbb42d : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1f3a26eea78c4f20ad4911cba042a0e3 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84151b66aa19439c8a8821b8175dda29 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3d103704d5d40e796886ba560d72041 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c110d4b5a84c2883793c16da0ae9d8 : SUCCESS in 4m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5258a5bf373444aa918bde2d4186f2bf : FAILURE in 3m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/46c854cde095460a902ecd50a5a548bf : SUCCESS in 4m 59s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/91edcf0caabc4979aeb38bc8fe26ecb5 : SUCCESS in 2h 30m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7c059d14344edd325f83a4587abe41fed18674ff","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-04 06:15:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e8b1ea8162637256e1b85113f9dc001705e89d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 08:47:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6aedabe6a45d4f7ab0faca10fbc42cb0 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/193aed8cdb644a279e9c744261845b85 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d420e6908564212bc0e2c025aa9a807 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0b44926334442dbabf4bf52311c5a89 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14f01b55507a40fd8967dd35baaec6c8 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41164f6a70b44e6b5946c61029f5977 : SUCCESS in 4m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d15ec6a405ee45f581fef52695cc81b8 : FAILURE in 3m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/554db5947f0d4258a3e88528c12fb363 : SUCCESS in 5m 02s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/98960d3454ce452aaef28dc4a19d2f6c : SUCCESS in 2h 25m 30s","accounts_in_message":[],"_revision_number":1},{"id":"68cc2fa55138d92dc44fcbf8de2ec7a992694e4e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-04 12:16:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3329b7404681aa88dda2b86926b2887a8959c3f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 14:42:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fa28a0ef6dc48d38d7c040f33d5e2f5 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ea76387ae7447449f06a07dc0b39cbb : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c85a51962a734bd198850858430ce641 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf87e155b5514464a5c183b0a58c1231 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8804232469b64d42bf1fc4d72bd4de2a : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99db644e2f814137b78be85566f2aaad : SUCCESS in 7m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dc12f203dca045e9bbd1391fffe07942 : SUCCESS in 6m 18s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/0150b98816bb4567a65dc917267fe83f : SUCCESS in 2h 19m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e01d67ead5ba53df92c199be91b6324934ae9cdb","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 17:13:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c652450deb0d900fa31d8c12aced054d43eb459","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-08 16:39:07.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0523ca2a3ed7fefeb0aedd761e38d3def21b62dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 16:39:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"402c5c1370b6b34ae8de26d2c526fa2fb54c3073","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 18:50:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2e028e6497b9d3ee01812d2dc12c7dfa3c8485ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 18:50:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff7dde9c6e0147f39fea6187f34f4beb : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cf700a30cd7474d8c3fda849f5cfc1c : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0b9a42d36aa4040a26f945afab43eae : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f64c88ec964945dbb2da8e08b87f765e : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50ec4530d11c4d30b9cfe73f612326e2 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd5a37b37dd4016bb331b03e93dc6ec : SUCCESS in 7m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fbc6859f18894fd98cbfe67bfc03ce49 : SUCCESS in 6m 01s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/2ca4c78b9345489e879ba1a31c6411b1 : SUCCESS in 2h 03m 25s","accounts_in_message":[],"_revision_number":1},{"id":"621b05695e0aea60a8d407d0e8e209f51fe79bd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 18:53:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc4774b3a424fedb26c8cb6c37a053f : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da7a1e275d7e8e590c32b7545e975a78ce6e8df6","revisions":{"da7a1e275d7e8e590c32b7545e975a78ce6e8df6":{"kind":"REWORK","_number":1,"created":"2019-10-03 14:35:36.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/16/686416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/16/686416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/16/686416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/16/686416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/16/686416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/16/686416/1"}}},"commit":{"parents":[{"commit":"4310bd2a06ea60b48930bb7e94276365fcaeb4fc","subject":"Merge \"PDF documentation build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4310bd2a06ea60b48930bb7e94276365fcaeb4fc"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2019-10-03 14:30:58.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2019-10-03 14:34:52.000000000","tz":0},"subject":"Change EAFP to LBYL style in nested transaction","message":"Change EAFP to LBYL style in nested transaction\n\nThere is a side effect that KeyError when looking up for existing\ntransaction stores a traceback to Python system. Later, if transaction\nfails and another exception is raised, the final traceback contains also\nthe original KeyError. Changing from EAFP [1] to LBYL [2] style avoids\nthis side-effect.\n\n[1] https://docs.python.org/3.4/glossary.html#term-eafp\n[2] https://docs.python.org/3.4/glossary.html#term-lbyl\n\nChange-Id: I30125ee023dfa59c4a5dde37195109e379e55c4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/da7a1e275d7e8e590c32b7545e975a78ce6e8df6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/da7a1e275d7e8e590c32b7545e975a78ce6e8df6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
