)]}'
{"id":"openstack%2Fovsdbapp~686507","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~I2d9599d7050a15f459e6845a112ffc9948b2a7bd","project":"openstack/ovsdbapp","branch":"stable/train","topic":"ovs-removed-python2-stable/train","hashtags":[],"change_id":"I2d9599d7050a15f459e6845a112ffc9948b2a7bd","subject":"Fix building OVS after python2 removal","status":"ABANDONED","created":"2019-10-03 22:20:31.000000000","updated":"2019-10-04 16:42:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f8d701f1dc42709fc179a1e01d99848510141eff","_number":686507,"virtual_id_number":686507,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-04 16:40:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"date":"2019-10-04 16:36:58.000000000","_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 22:26:02.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-10-04 16:36:58.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-10-04 16:40:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77b21ca3dd9a6895b33ebe86638fcf2167c49dc2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-03 22:20:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"df084972a2dc1452901f47a8ce844d66600d6810","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-03 22:26:02.000000000","message":"Patch Set 1: Code-Review-1\n\nThis should be -Workflow until the master change is merged.  Also, there should be a \"cherry picked from commit\" line in the commit message.","accounts_in_message":[],"_revision_number":1},{"id":"7b76f63dd9500ee14949125f66cabffafb827e3e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-03 22:32:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"62bdc52354fd3a31671dce3615dbd8cc6f8f90da","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-03 22:34:32.000000000","message":"Patch Set 2: Workflow-1\n\nThe cherry-pick button doesn\u0027t add the cherry-picked comment unless the patch has merged, so I manually added it despite the fact that it could technically change. Just trying to get it so that some other patches could get started testing while waiting for the other to merge. I\u0027ll add W-1, but since usually only 2 people ever approve patches and often only one for simple patches like this and the other one is asleep right now, it won\u0027t have much of an actual effect. :D","accounts_in_message":[],"_revision_number":2},{"id":"477be5f3e162c6ab2727866847b810a88d4015d6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-03 22:42:19.000000000","message":"Patch Set 2:\n\nLooks like I\u0027m going to have to squash the two patches to get functional tests to work... :(","accounts_in_message":[],"_revision_number":2},{"id":"6cf50bd4728066328c65fc297b3f9307d5e40af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 22:43:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53fc25d8ec014d5398e2e01e2a0e6757 : SUCCESS in 3m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2333e5ad1f274fd9be7235b6d3e50bf7 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b72f29396f3d4d7b98ad03e56c4d1616 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/107d850e14c642d187964d9da8926783 : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4eca847bc8814e62afc3e66244da492c : SUCCESS in 4m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1cef3f68abfd49e780f9362791ca9c76 : FAILURE in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"4b59b98210be64b5384b6fb6c7c579db100c01af","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 13:30:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"278d7371477df7cfee70920ae87f9703e88c0d76","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:26:55.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"c51b74cef89ad7bccb88a740c3a48f04a61b647a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:28:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"bd8612c80566322c2eadc8365d6d402d27005135","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-04 16:36:58.000000000","message":"Patch Set 3:\n\nthis can be abandoned","accounts_in_message":[],"_revision_number":3},{"id":"b6b7903b9ac98095005cd8ad07cf5e5dd2b066eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 16:40:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be2a1d5b2b7d40edabc7127ba8fccc8a : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a85dfaac0146409185530ef8709cb904 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d6743273056416391e602ab87931fcf : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/705eb1e70e8f4a27a16dd98c51942b73 : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a11da4dea0e74ad68d020388b414a47e : SUCCESS in 4m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7645b762d2644944ba6d5ce3d9f0137e : FAILURE in 5m 55s","accounts_in_message":[],"_revision_number":3},{"id":"b7fc76c541b29e744e2e7d8297ae9be7d9233044","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:42:20.000000000","message":"Abandoned\n\npart of squash here https://review.opendev.org/#/c/686238/","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7574897bc0474aef850c9822c582cebeab882da5","revisions":{"c9ffbfde7ee639ebe59175ab65728f56d1f3b099":{"kind":"REWORK","_number":1,"created":"2019-10-03 22:20:31.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/07/686507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/07/686507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/1"}}},"commit":{"parents":[{"commit":"74b52bb3491a9c4f6ec40e15d74652fc415269e0","subject":"Add HA Chassis Group related commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/74b52bb3491a9c4f6ec40e15d74652fc415269e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-03 21:09:47.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-03 22:20:31.000000000","tz":0},"subject":"Fix building OVS after python2 removal","message":"Fix building OVS after python2 removal\n\nSwitch functional to use python3 to install python3-based\ndependencies, which the OVS build checks for. Also, remove\nthe py36 test\n\nChange-Id: I2d9599d7050a15f459e6845a112ffc9948b2a7bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c9ffbfde7ee639ebe59175ab65728f56d1f3b099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c9ffbfde7ee639ebe59175ab65728f56d1f3b099"}]},"branch":"refs/heads/stable/train"},"bf4ec880f5a613c76557235bbd0331bdcbeaf760":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-03 22:32:12.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/07/686507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/07/686507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/2"}}},"commit":{"parents":[{"commit":"74b52bb3491a9c4f6ec40e15d74652fc415269e0","subject":"Add HA Chassis Group related commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/74b52bb3491a9c4f6ec40e15d74652fc415269e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-03 21:09:47.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-03 22:31:56.000000000","tz":0},"subject":"Fix building OVS after python2 removal","message":"Fix building OVS after python2 removal\n\nSwitch functional to use python3 to install python3-based\ndependencies, which the OVS build checks for. Also, remove\nthe py36 test\n\nChange-Id: I2d9599d7050a15f459e6845a112ffc9948b2a7bd\n(cherry-picked from commit d5420ab5d248b93c5ce1c65d29521dc52ad3c5ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bf4ec880f5a613c76557235bbd0331bdcbeaf760"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bf4ec880f5a613c76557235bbd0331bdcbeaf760"}]},"branch":"refs/heads/stable/train"},"7574897bc0474aef850c9822c582cebeab882da5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-10-04 16:28:48.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/07/686507/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/07/686507/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/07/686507/3"}}},"commit":{"parents":[{"commit":"e8c5763179d357c060cb4018ccdd7075ebb8c7ae","subject":"Update gitreview for train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e8c5763179d357c060cb4018ccdd7075ebb8c7ae"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-03 21:09:47.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 16:28:48.000000000","tz":0},"subject":"Fix building OVS after python2 removal","message":"Fix building OVS after python2 removal\n\nSwitch functional to use python3 to install python3-based\ndependencies, which the OVS build checks for. Also, remove\nthe py36 test\n\nChange-Id: I2d9599d7050a15f459e6845a112ffc9948b2a7bd\n(cherry-picked from commit d5420ab5d248b93c5ce1c65d29521dc52ad3c5ae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7574897bc0474aef850c9822c582cebeab882da5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7574897bc0474aef850c9822c582cebeab882da5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
