)]}'
{"id":"openstack%2Fovsdbapp~686766","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~Ieab25244e3d1bff56537de1d5915736f3e8c26dc","project":"openstack/ovsdbapp","branch":"stable/train","hashtags":[],"change_id":"Ieab25244e3d1bff56537de1d5915736f3e8c26dc","subject":"Update gitreview for train","status":"ABANDONED","created":"2019-10-04 16:20:23.000000000","updated":"2019-10-04 16:43:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dd920750e287df2c89363adba873214a358bdfba","_number":686766,"virtual_id_number":686766,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-04 16:27:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 16:27:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8bdf9d1b24a54965ca5bb247a6705dddc52d643b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:20:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ca3b521a0744ebe7f736d1c9ced9daa08c121f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 16:27:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d23fdecb034a8fa0c4cb2b22564b0b : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d5d7a663fe0414f920d4c23dabb8d17 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/974eaedea32c4acf956fb4fcfb85ef27 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d656b6fc073483292bcb77436f6d3b9 : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b34da3a52304925a45dcbc72256694d : SUCCESS in 5m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/011587b6c7244df79884c8f0c93c6e9d : FAILURE in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2a37e4c1efc94da6ae8d05d2538af42f : FAILURE in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"dd920750e287df2c89363adba873214a358bdfba","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 16:43:02.000000000","message":"Abandoned\n\npart of squash here https://review.opendev.org/#/c/686238/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8c5763179d357c060cb4018ccdd7075ebb8c7ae","revisions":{"e8c5763179d357c060cb4018ccdd7075ebb8c7ae":{"kind":"REWORK","_number":1,"created":"2019-10-04 16:20:23.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/66/686766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/66/686766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/66/686766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/66/686766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/66/686766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/66/686766/1"}}},"commit":{"parents":[{"commit":"74b52bb3491a9c4f6ec40e15d74652fc415269e0","subject":"Add HA Chassis Group related commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/74b52bb3491a9c4f6ec40e15d74652fc415269e0"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 16:19:39.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 16:19:39.000000000","tz":-300},"subject":"Update gitreview for train","message":"Update gitreview for train\n\nChange-Id: Ieab25244e3d1bff56537de1d5915736f3e8c26dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e8c5763179d357c060cb4018ccdd7075ebb8c7ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e8c5763179d357c060cb4018ccdd7075ebb8c7ae"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
