)]}'
{"id":"openstack%2Fovsdbapp~686848","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~I427f5c87058de04db21b23b26d5f1070b2efeb61","project":"openstack/ovsdbapp","branch":"stable/train","hashtags":[],"change_id":"I427f5c87058de04db21b23b26d5f1070b2efeb61","subject":"Handle unset OVN_SRCDIR","status":"MERGED","created":"2019-10-04 22:34:13.000000000","updated":"2019-10-08 08:50:17.000000000","submitted":"2019-10-07 22:01:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686848-1570485684421-70f2cce9","meta_rev_id":"73698ace6d31f2839e527bc7e46e669aa9ba92ec","_number":686848,"virtual_id_number":686848,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-07 22:01:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2019-10-07 19:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2019-10-07 19:38:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 22:39:40.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2019-10-04 22:39:52.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-10-07 22:01:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2942b97abb8777627f79969eb5a80d42ec365712","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 22:34:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f9bfb88f267d83617645ec0698ca86d1a811b09","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 22:40:36.000000000","message":"Patch Set 1:\n\noops, this should have gone to master first...","accounts_in_message":[],"_revision_number":1},{"id":"c0e08156ba911e43a5e8fb7f9b5c00909d2cf997","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 22:43:59.000000000","message":"Patch Set 1: Workflow-1\n\nwaiting for master","accounts_in_message":[],"_revision_number":1},{"id":"5ec98fb595a386d4cbf306c49ef66f2b8b1ab5df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 22:46:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58fa8203c82f4f3ba224e3a5ff4bb861 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/093129231404455cba9a6cc8795c55c0 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb464dfb7c8c4220a0f521607588c8f1 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a8e7a6f44baf4c1a868069551f679e18 : SUCCESS in 5m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83331163e29448eca1474e6ba20ee146 : SUCCESS in 4m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/41a38f61be874b9fb9878c091f0fbe25 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7a66184002f9cbca2ef1257babf28be9e9e35476","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-04 22:46:16.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"390ac4dc81afd820938aadfb757a84e94cfdff35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 22:57:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b30b1b6a9ad24ee3ade4705e9bd99530 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b180fe579ed243d99bc5a60dca1f95cd : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f09efffd8dd43b298e375088bd85c66 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04133702d96f41d7bfb3a2558a7ca937 : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8359fc7273b449c6b617d313e7195f1d : SUCCESS in 3m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a6c592caebd043dd9ad4c0a565c76757 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"85cb70c89f447aaf66836fe278b50997b0724ad3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-07 19:38:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nClean cherry-pick, self-approving so release can be made.","accounts_in_message":[],"_revision_number":2},{"id":"090ee7d24be907c2920019259319b5a15c4ab48f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 19:38:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3590fb90d172654c3fc04c801e10325171de676","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 22:01:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"30239073981cacc3430ec497213dba6cb98bffd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 22:01:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7896ca7ed45e4a7198c14fbfa9ee8f03 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea6ceb5e5853466da5a5c4607705664d : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afe933afa3a6482887de6c7b59157abc : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4521cba0625497f9b68317160f56989 : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e0dd5cad1d14d939b94a602ba04d6fe : SUCCESS in 5m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ccb07a2bef764d208b4376e618a188ff : SUCCESS in 6m 26s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/8436d8c4f2e24d40a531626a74f97b57 : SUCCESS in 2h 16m 25s","accounts_in_message":[],"_revision_number":2},{"id":"73698ace6d31f2839e527bc7e46e669aa9ba92ec","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-08 08:50:17.000000000","message":"Patch Set 2:\n\nI also approve :) +2","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"206cf1465c7fb4f6a5538bca0fedd63425c6ff46","revisions":{"df194b173ecca1b085308bced72d2d0867a72e3e":{"kind":"REWORK","_number":1,"created":"2019-10-04 22:34:13.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/48/686848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/48/686848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/686848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/686848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/686848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/48/686848/1"}}},"commit":{"parents":[{"commit":"251d7c8f17864aaa3821d2f2670d60eafba5e95c","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/251d7c8f17864aaa3821d2f2670d60eafba5e95c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 22:33:01.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 22:34:12.000000000","tz":-300},"subject":"Handle unset OVN_SRCDIR","message":"Handle unset OVN_SRCDIR\n\nAllow the usage of system-installed OVN. This will allow tests that\nuse ovsdbapp\u0027s venv.py in networking-ovn to pass.\n\nChange-Id: I427f5c87058de04db21b23b26d5f1070b2efeb61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/df194b173ecca1b085308bced72d2d0867a72e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/df194b173ecca1b085308bced72d2d0867a72e3e"}]},"branch":"refs/heads/stable/train"},"206cf1465c7fb4f6a5538bca0fedd63425c6ff46":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-04 22:46:16.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/48/686848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/48/686848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/686848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/686848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/48/686848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/48/686848/2"}}},"commit":{"parents":[{"commit":"251d7c8f17864aaa3821d2f2670d60eafba5e95c","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/251d7c8f17864aaa3821d2f2670d60eafba5e95c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 22:33:01.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-10-04 22:45:13.000000000","tz":0},"subject":"Handle unset OVN_SRCDIR","message":"Handle unset OVN_SRCDIR\n\nAllow the usage of system-installed OVN. This will allow tests that\nuse ovsdbapp\u0027s venv.py in networking-ovn to pass.\n\nChange-Id: I427f5c87058de04db21b23b26d5f1070b2efeb61\n(cherry picked from commit 0e3d9743c06ce1112a96bf54d08090365fedf9b9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/206cf1465c7fb4f6a5538bca0fedd63425c6ff46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/206cf1465c7fb4f6a5538bca0fedd63425c6ff46"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
