)]}'
{"id":"openstack%2Fovsdbapp~690861","triplet_id":"openstack%2Fovsdbapp~master~I579f7297be2b86666e9b494ae2d483454dab4cf1","project":"openstack/ovsdbapp","branch":"master","topic":"ussuri","hashtags":[],"change_id":"I579f7297be2b86666e9b494ae2d483454dab4cf1","subject":"Switch to Ussuri jobs","status":"MERGED","created":"2019-10-24 06:48:19.000000000","updated":"2019-10-28 19:15:44.000000000","submitted":"2019-10-28 19:13:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690861-1572290003708-b61b701d","meta_rev_id":"a0ab199caa03b55ab49c859942cbe418edbbf9a7","_number":690861,"virtual_id_number":690861,"owner":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-28 19:13:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":2,"date":"2019-10-28 17:09:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-25 07:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":1,"date":"2019-10-28 17:09:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 07:16:43.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2019-10-28 17:09:22.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2019-10-28 19:13:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a4205f0d26d1bb931b3360d61db5d2aedd331ec","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-10-24 06:48:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3ff7479da644dd426b54e9dc55db7142d40c7ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 09:43:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae424140ce4746429043c8d7c0837e39 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a919ae2030b474c8fc455a84199a89a : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a329a933bec147fb81331c4ea0222409 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b123e50b352471082eb30c30c612cde : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cdebd2212cd46c7bbd6a7360c996b09 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d419d85251b84fde8ea994315dfdfffa : SUCCESS in 6m 07s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/bf75cd80a54b4b5db1abe49bb86c4c1e : SUCCESS in 5m 39s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/2f910877a2854d5da0c479cb3d60220d : SUCCESS in 2h 26m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ee1ac2fb54156409ce5e2e573e90f1653647974d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2019-10-25 07:16:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3cf2b5a102dabdf5bbbacd938a866e192f0b5cb","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-10-28 17:09:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0fb1eae86f3b8629dd409287fa17e5669fc57b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 17:09:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98d2613563b35fb7befac71a4c91f20349d1cda0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:13:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0bde8917f7b75b31b32097002b4ffe14084087b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:13:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8874abccd0b463f819149e0cc243cfe : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a463b31977f1461ab6f5ddea64ddedc1 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/710ee0127d8449f7b0fe2d3f0bade691 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/348f752b4e6749b2b893a0cdedfa5973 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f64b11c8d6c48e7b6ddf7bdd02d23c9 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a37b9db834cb470f95fd42532555eedb : SUCCESS in 4m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3512e064b3be40e9aaf7b875b1534556 : SUCCESS in 5m 28s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/88d39ac3bd5f4fbfa3abe5f59bd1be48 : SUCCESS in 2h 02m 40s","accounts_in_message":[],"_revision_number":1},{"id":"052225e8c72e2cb0e8cb32e928f0ce6a12fab859","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 19:15:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6aff74062f4f5d98976cfa100a4d35 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2a04b627baa4bcc26e1c593785c99bd82024622","revisions":{"c2a04b627baa4bcc26e1c593785c99bd82024622":{"kind":"REWORK","_number":1,"created":"2019-10-24 06:48:19.000000000","uploader":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"ref":"refs/changes/61/690861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/61/690861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/690861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/690861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/61/690861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/61/690861/1"}}},"commit":{"parents":[{"commit":"66510183a1ef7ed2ada30f9dde4c28a413e3594f","subject":"ovn_northbound: make all list commands ReadOnlyCommand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/66510183a1ef7ed2ada30f9dde4c28a413e3594f"}]}],"author":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-24 06:48:45.000000000","tz":480},"committer":{"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","date":"2019-10-24 06:48:45.000000000","tz":480},"subject":"Switch to Ussuri jobs","message":"Switch to Ussuri jobs\n\nChange-Id: I579f7297be2b86666e9b494ae2d483454dab4cf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c2a04b627baa4bcc26e1c593785c99bd82024622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c2a04b627baa4bcc26e1c593785c99bd82024622"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
