)]}'
{"id":"openstack%2Fovsdbapp~692941","triplet_id":"openstack%2Fovsdbapp~master~Ia710b8f394e9446f0da83221e314d73d2edcfb08","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"Ia710b8f394e9446f0da83221e314d73d2edcfb08","subject":"Fix debug_venv for ovn split","status":"MERGED","created":"2019-11-05 02:53:15.000000000","updated":"2019-11-08 23:23:53.000000000","submitted":"2019-11-08 23:22:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692941-1573255348769-4445955c","meta_rev_id":"6d580230ce1ae6b445a5a0c7660002173eb708e8","_number":692941,"virtual_id_number":692941,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-08 23:22:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2019-11-07 04:23:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-11-08 20:57:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-07 04:23:35.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-11-08 23:22:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9fc1e561f21415e7c6d28ec83fcb043856183a5","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-11-05 02:53:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"928d724c582ec02e297bddfd13eab9e509c96dcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 03:02:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/845b7bb37d6e495db69546a8d3cedbd1 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e9d598e65245d9bddeba44d32e2614 : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/396a515fc7e045209e0959899c13743c : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/352e60b2bad9448bb06bb864dd3c9d22 : SUCCESS in 2m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65b420b8ff2c4e7a8328b5a710984ce4 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7591375c4f8049a99d4e1cefa3d76d59 : SUCCESS in 4m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8d5f1cf2f49c4b018fee56521de57144 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b670d8744e5828c0ab5944df9eb1a8eba375bbc7","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-11-06 00:07:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"751cd8cc5b0c2d0ab4b7d37eb39b91b411ae2ea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 02:03:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e13dfd7578c3483f8b47c45fa412e909 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a9f8e3739642c5878a85b3cfea7df0 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/13753cb572164564a84cc97ded90671c : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c1e9565373145b5956e49f13bac76bd : SUCCESS in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fa1542d02ff4af784a65a4c522c80e6 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c4d80fb4b54f3ab538c0698205dfc9 : SUCCESS in 5m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/11eed5570fd44d64aa0b6ad9a4e27484 : SUCCESS in 5m 30s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/999717bfa6294ec8b8ee4da1c58c5603 : SUCCESS in 1h 53m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a743899715eaf896d43c694ba621651e094878ef","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-07 04:23:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6c73c596f3f9c79884a0727cad2c5077216ff569","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2019-11-08 20:57:46.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aac325b3a858b19c83cc84153b2d2f3bcf414d1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 20:57:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"91039642bdbd8c3c07bbda67d945b3b12f1384a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 23:22:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48c40b0d7c4940d78ba44047782295b6 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eed531e4b4b947098c987b3b586ace89 : SUCCESS in 2m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52d51323983a404eb16085be2cf531c9 : SUCCESS in 2m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b3f7d79662d4ca783eae98604a111fa : SUCCESS in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2300f3cbb19f42ff85775708718f37a1 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd4e36b0e7c242168615fd1a72c079c9 : SUCCESS in 4m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/388a9d8676a144a184ca0186e755b19a : SUCCESS in 4m 38s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/77970269080243699639b826fe5d9668 : SUCCESS in 2h 24m 03s","accounts_in_message":[],"_revision_number":2},{"id":"73dcda5ec6c6b2ee641331f003860a1a743f46fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 23:22:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6d580230ce1ae6b445a5a0c7660002173eb708e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 23:23:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1c1ad5791ea490d98b1f9377355de57 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0bd24c269d07e1181d9e6a7ed4267a63d7ad1921","revisions":{"86711a05b4b88711669236887735ac49ba9b969d":{"kind":"REWORK","_number":1,"created":"2019-11-05 02:53:15.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/41/692941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/692941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/692941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/692941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/692941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/692941/1"}}},"commit":{"parents":[{"commit":"5a61df8407abb0dc165568972f4ec05dbf84f48c","subject":"Merge \"Add consts for OVN install path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5a61df8407abb0dc165568972f4ec05dbf84f48c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-11-05 02:52:37.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-11-05 02:53:14.000000000","tz":-360},"subject":"Fix debug_venv for ovn split","message":"Fix debug_venv for ovn split\n\nChange-Id: Ia710b8f394e9446f0da83221e314d73d2edcfb08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/86711a05b4b88711669236887735ac49ba9b969d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/86711a05b4b88711669236887735ac49ba9b969d"}]},"branch":"refs/heads/master"},"0bd24c269d07e1181d9e6a7ed4267a63d7ad1921":{"kind":"REWORK","_number":2,"created":"2019-11-06 00:07:15.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/41/692941/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/41/692941/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/692941/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/692941/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/41/692941/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/41/692941/2"}}},"commit":{"parents":[{"commit":"5a61df8407abb0dc165568972f4ec05dbf84f48c","subject":"Merge \"Add consts for OVN install path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/5a61df8407abb0dc165568972f4ec05dbf84f48c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-11-05 02:52:37.000000000","tz":-360},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-11-06 00:07:14.000000000","tz":-360},"subject":"Fix debug_venv for ovn split","message":"Fix debug_venv for ovn split\n\nChange-Id: Ia710b8f394e9446f0da83221e314d73d2edcfb08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0bd24c269d07e1181d9e6a7ed4267a63d7ad1921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0bd24c269d07e1181d9e6a7ed4267a63d7ad1921"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
