)]}'
{"id":"openstack%2Fovsdbapp~713067","triplet_id":"openstack%2Fovsdbapp~stable%2Fstein~Ice22858df465b37b973718be718673aab46193ce","project":"openstack/ovsdbapp","branch":"stable/stein","hashtags":[],"change_id":"Ice22858df465b37b973718be718673aab46193ce","subject":"Make Event logging more useful by default","status":"MERGED","created":"2020-03-13 21:57:33.000000000","updated":"2020-04-06 17:36:15.000000000","submitted":"2020-04-06 17:36:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713067-1586194575852-e8a3316c","meta_rev_id":"a0485f10e092c5218750ba5958bfa53104a03471","_number":713067,"virtual_id_number":713067,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-06 17:36:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":2,"date":"2020-04-06 15:15:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-04-03 13:11:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":1,"date":"2020-04-06 15:15:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 21:58:34.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-04-03 13:11:49.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-06 15:15:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-06 17:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8e6cd0d11c1c10bdaa602c4e109c9c4e98abae7","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-13 21:57:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36f86829e186e593ac15947434f17d4690d6e8a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 00:06:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02813c530d61478fbe1f37a3b5959f29 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74386c46780946c9ad9cb226ef25568a : SUCCESS in 3m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ff4b1839eea04b2f9e04ba2ac58debae : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1cfa7bac3494dfe964f1a6d692e1e10 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5cdbaf2cdb346479e7519ebf8c45384 : SUCCESS in 6m 51s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/332f8991dfc74062baa2ee1e17a6c358 : FAILURE in 7m 35s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/f701f6a21eeb4e7aac3f02bb62edb3cc : SUCCESS in 2h 08m 28s","accounts_in_message":[],"_revision_number":1},{"id":"25e523a5c7993ee90ce88f6036b926767b5371cf","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 08:51:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"be96f7924b616c66fde2989cef1d8de95675b074","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 10:55:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fa38456009e4f6da975f5b26106352c : SUCCESS in 7m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a96aab5948434d6f9bdd161b3c254e13 : SUCCESS in 7m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c59017c775974531974671107c124107 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0137a9efb4224f5e847fc3bd39b13cf7 : SUCCESS in 7m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27004d406218485c881b498d9a325d6a : SUCCESS in 7m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a37863dfd5704cfcaf7015be936d18cd : SUCCESS in 8m 04s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/abe83affdbc3415293958ab1370b6105 : SUCCESS in 1h 58m 42s","accounts_in_message":[],"_revision_number":2},{"id":"49841fd0e2097ca05f0ac016dd7634f9a9cdf3bd","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-03 13:11:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fe79a727207b5c4161e578dc7b59bb8ad0daea0a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-06 15:15:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b5cf625d248337dff0ad54d06738462c4c825a25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 15:15:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e236a2236c2766831310e944ed4ef22dba626713","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 17:36:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8955f856e9674ce18009b3fa68de3733 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/36dca18d01484e1b99e39e3d6323757b : SUCCESS in 7m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/093fa0a2b35342ea92e100fc24e42a77 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5513cd25af0a46acb3d5e13b0918849d : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28757bc26a7c4174bb904874e3e2bab1 : SUCCESS in 5m 31s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d13b9bb57f0a407da0e9bf0ec78dfd01 : SUCCESS in 7m 16s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/fc664d12a3f4439bb83717be24b86ddd : SUCCESS in 2h 17m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a0485f10e092c5218750ba5958bfa53104a03471","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 17:36:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9cb971ef46f108e1f1ca9a80ddb92b8c01e4e2ad","revisions":{"d896bb162d72d03d2b489942623429e99012998b":{"kind":"REWORK","_number":1,"created":"2020-03-13 21:57:33.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/67/713067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/67/713067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/713067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/713067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/713067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/67/713067/1"}}},"commit":{"parents":[{"commit":"07b3b48fa42727a99e57f227b9b42cb3044e54f8","subject":"Fix functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/07b3b48fa42727a99e57f227b9b42cb3044e54f8"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-05-21 20:12:12.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-13 21:56:10.000000000","tz":-240},"subject":"Make Event logging more useful by default","message":"Make Event logging more useful by default\n\nLog messages like:\n\nChassisGatewayAgentEvent : Matched Chassis, create, None None\n\nare not particularly useful. This adds a default repr to RowEvents\nand also a Row stringifier (which eventually could go in the ovs\nlib). Subclasses of RowEvent that rely on custom match functions\nshould write their own __repr__ methods.\n\nChange-Id: Ice22858df465b37b973718be718673aab46193ce\n(cherry picked from commit ceea8a23f1e113d46da8b62ecf1cb77acbee24a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d896bb162d72d03d2b489942623429e99012998b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d896bb162d72d03d2b489942623429e99012998b"}]},"branch":"refs/heads/stable/stein"},"9cb971ef46f108e1f1ca9a80ddb92b8c01e4e2ad":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-03 08:51:03.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/67/713067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/67/713067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/713067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/713067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/67/713067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/67/713067/2"}}},"commit":{"parents":[{"commit":"b22e1c4e43ca4f095bff8cb228ef83dd1474a403","subject":"Unbreak mock-releated autospec issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b22e1c4e43ca4f095bff8cb228ef83dd1474a403"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-05-21 20:12:12.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-04-03 08:51:03.000000000","tz":0},"subject":"Make Event logging more useful by default","message":"Make Event logging more useful by default\n\nLog messages like:\n\nChassisGatewayAgentEvent : Matched Chassis, create, None None\n\nare not particularly useful. This adds a default repr to RowEvents\nand also a Row stringifier (which eventually could go in the ovs\nlib). Subclasses of RowEvent that rely on custom match functions\nshould write their own __repr__ methods.\n\nChange-Id: Ice22858df465b37b973718be718673aab46193ce\n(cherry picked from commit ceea8a23f1e113d46da8b62ecf1cb77acbee24a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9cb971ef46f108e1f1ca9a80ddb92b8c01e4e2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9cb971ef46f108e1f1ca9a80ddb92b8c01e4e2ad"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
