)]}'
{"id":"openstack%2Fovsdbapp~713068","triplet_id":"openstack%2Fovsdbapp~stable%2Frocky~Ice22858df465b37b973718be718673aab46193ce","project":"openstack/ovsdbapp","branch":"stable/rocky","hashtags":[],"change_id":"Ice22858df465b37b973718be718673aab46193ce","subject":"Make Event logging more useful by default","status":"ABANDONED","created":"2020-03-13 22:03:22.000000000","updated":"2020-04-06 15:11:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c74b2269624fe871441a966e60a97d59c136908","_number":713068,"virtual_id_number":713068,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-03 10:13:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2020-04-06 06:20:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 22:03:22.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-03 10:13:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-06 06:20:02.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"}],"messages":[{"id":"abd486d784db7c82a62a1b43bc163edb4d283d39","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-13 22:03:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c387cf3a9f24e9085167e044afef4a268d13012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:41:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/910a548b02f24f0c9248331227a1bdb8 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7497e9afe4d43a99cc78c9926568065 : SUCCESS in 3m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/719b0a10c1534154b81dc16c75f5215e : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37f18a040f5b401cb6875d597a83ffee : SUCCESS in 5m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/662fb1d90dac4bb5be3bcc14dd058d53 : SUCCESS in 5m 27s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/038561ef398d4aceb65d9cc71efe5c88 : FAILURE in 32m 49s","accounts_in_message":[],"_revision_number":1},{"id":"80d7a97bf0ab46066172831f4f9baf54ef12db09","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 09:08:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0650a748a933a9a4820e531b0c2b784da2c492b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 10:13:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aca5dca2c9804a808b3c00acbf3a85da : SUCCESS in 8m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/caa805383781480da0ac4285741a1f60 : SUCCESS in 6m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/55da960f4c734139930f22fda7eb0077 : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d73828ce759418d9836f13f5cc2aed8 : SUCCESS in 10m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0134209302e7464bae7860434a38b357 : SUCCESS in 7m 57s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/1ad62e7bff8a493ca5400ee676059b3c : FAILURE in 42m 08s","accounts_in_message":[],"_revision_number":2},{"id":"00fbe04544093ca3ef84fa9a5ff3d52447d2681a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-04-06 06:20:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2c74b2269624fe871441a966e60a97d59c136908","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-06 15:11:38.000000000","message":"Abandoned\n\nAbandoning support for these changes on stable/rocky.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b2b60f42af47432c6851b1a56bc49de99d371be6","revisions":{"7e7f8be5d734f75a028380d53af589dd77088ab5":{"kind":"REWORK","_number":1,"created":"2020-03-13 22:03:22.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/68/713068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/68/713068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/68/713068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/68/713068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/68/713068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/68/713068/1"}}},"commit":{"parents":[{"commit":"2b1552d8b044750afab5a0e43b5354b3b1d3e28d","subject":"Expand retry behavior to cover other python-ovs methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2b1552d8b044750afab5a0e43b5354b3b1d3e28d"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-05-21 20:12:12.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-13 22:02:52.000000000","tz":-240},"subject":"Make Event logging more useful by default","message":"Make Event logging more useful by default\n\nLog messages like:\n\nChassisGatewayAgentEvent : Matched Chassis, create, None None\n\nare not particularly useful. This adds a default repr to RowEvents\nand also a Row stringifier (which eventually could go in the ovs\nlib). Subclasses of RowEvent that rely on custom match functions\nshould write their own __repr__ methods.\n\nChange-Id: Ice22858df465b37b973718be718673aab46193ce\n(cherry picked from commit ceea8a23f1e113d46da8b62ecf1cb77acbee24a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7e7f8be5d734f75a028380d53af589dd77088ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7e7f8be5d734f75a028380d53af589dd77088ab5"}]},"branch":"refs/heads/stable/rocky"},"b2b60f42af47432c6851b1a56bc49de99d371be6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-03 09:08:21.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/68/713068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/68/713068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/68/713068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/68/713068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/68/713068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/68/713068/2"}}},"commit":{"parents":[{"commit":"45057e3fb1e9c1cf3c66382e7cf21c5ac5f2b136","subject":"Unbreak mock-releated autospec issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/45057e3fb1e9c1cf3c66382e7cf21c5ac5f2b136"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-05-21 20:12:12.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-04-03 09:08:21.000000000","tz":0},"subject":"Make Event logging more useful by default","message":"Make Event logging more useful by default\n\nLog messages like:\n\nChassisGatewayAgentEvent : Matched Chassis, create, None None\n\nare not particularly useful. This adds a default repr to RowEvents\nand also a Row stringifier (which eventually could go in the ovs\nlib). Subclasses of RowEvent that rely on custom match functions\nshould write their own __repr__ methods.\n\nChange-Id: Ice22858df465b37b973718be718673aab46193ce\n(cherry picked from commit ceea8a23f1e113d46da8b62ecf1cb77acbee24a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b2b60f42af47432c6851b1a56bc49de99d371be6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b2b60f42af47432c6851b1a56bc49de99d371be6"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
