)]}'
{"id":"openstack%2Fovsdbapp~713069","triplet_id":"openstack%2Fovsdbapp~stable%2Fqueens~Ice22858df465b37b973718be718673aab46193ce","project":"openstack/ovsdbapp","branch":"stable/queens","hashtags":[],"change_id":"Ice22858df465b37b973718be718673aab46193ce","subject":"Make Event logging more useful by default","status":"MERGED","created":"2020-03-13 22:03:46.000000000","updated":"2020-04-08 20:17:21.000000000","submitted":"2020-04-08 20:17:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713069-1586377041674-30de0cd9","meta_rev_id":"c089cdba72cfdd08401ec7b18c02f6be5f7e6d9b","_number":713069,"virtual_id_number":713069,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-08 17:33:43.000000000","_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 20:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-04-08 15:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-04-08 17:19:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-04-08 17:19:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 22:05:59.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-04-08 15:43:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-08 17:19:32.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-08 20:17:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73e46cc976432456918c63843125d576c28e2541","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-13 22:03:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a7b59a58d7ee23d06b11a5d35d024c0f143011f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 23:03:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb0c91249f2b441382a04e3ccce5e974 : FAILURE in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec2c65a9376b45c090bd2a0be420abdb : SUCCESS in 4m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/454237784b7c405b8b41b8264aac80b2 : SUCCESS in 4m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/7957fd994cc14370be90d859b2def02c : SUCCESS in 6m 42s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/a2691262dce045a2bdb62b443988442b : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"96f082064eb2059cc9ab93071f288cc9f87094cf","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 09:10:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ef4005b4f69093f21a19f0c7f7d03da4a59bb323","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 10:44:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0fdb70e84a42e0b18ff98ee78762c3 : FAILURE in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e9ad97e9708484ab42c00bc949ba7ae : SUCCESS in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a045a3adfc0348f4aaca0f4e58be93c2 : SUCCESS in 6m 52s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/117bcbb21e3e4afc83ece1d178c2fddb : SUCCESS in 8m 00s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/54974ba7532b4c588268b7b9aa5f97d2 : SUCCESS in 1h 04m 48s\n\nWarning:\n  Comments left for invalid file tools/debug_venv.py","accounts_in_message":[],"_revision_number":2},{"id":"20eae20c2fa28e7286a604744a364be1543676c2","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 20:30:13.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"785468bbfd197a4dfc19b03e1727ab0f6c9bc802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 21:33:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0584ea3a44b4b30942039c85073ec41 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a6a2533b6c784994b82dc6b9c488c583 : SUCCESS in 7m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/700c0bda13604b7d91770b0a48de1a8b : SUCCESS in 7m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/86474035b684451f98c7df420e611b8a : SUCCESS in 7m 03s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/489b2a68ce4348908ca6eef0e487d489 : SUCCESS in 1h 00m 10s\n\nWarning:\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.5/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional/lib/python3.5/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-py35: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py35/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-py35: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py35/lib/python3.5/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py27/lib/python2.7/site-packages/setuptools/dist.py in the commit\u0027\n  Comments left for invalid file .tox/py35/lib/python3.5/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/pep8/lib/python3.5/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/py35/lib/python3.5/site-packages/pep8.py\n  Comments left for invalid file .tox/functional/lib/python3.5/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/functional/lib/python3.5/site-packages/pep8.py\n  Comments left for invalid file .tox/py27/lib/python2.7/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/pep8/lib/python3.5/site-packages/pep8.py","accounts_in_message":[],"_revision_number":3},{"id":"32b6600a8b969432a803da9f72be2e3de60ee923","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-07 18:37:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c8549052061ace42d62a1903972c9d6fa5005880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 20:16:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53744cedb0c4f38a42a10efe0eeb936 : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/20d79d701e5548e5a09a41e285e5df64 : SUCCESS in 7m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bca9c99e220e47b8bce1c7490cffdae7 : SUCCESS in 7m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a6527f77e38342fe93ae0274a704b2f3 : SUCCESS in 7m 34s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/adfc14ac374244b095d464d1c358896d : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":4},{"id":"a341e2b197ef25b73c13a43d8a2d60f1a3559a02","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-08 15:43:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7fc59e8b3adf9f04067a9f0fc81c80a426e930da","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-08 17:19:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nthanks flavio!","accounts_in_message":[],"_revision_number":4},{"id":"83c4816e97845e9a8768f7570ab20183f0b31b26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:19:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a0f044a5c7e83f9e48ccbc6eb66b0553efe28ae9","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-08 17:33:43.000000000","message":"Patch Set 4:\n\n\u003e thanks flavio!\n\nmy pleasure, bud!","accounts_in_message":[],"_revision_number":4},{"id":"21b271f0086680aff920896f7fc4b1e3cb8d1a26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:17:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb3fbee7ce84504a3191c9f01b05191 : SUCCESS in 7m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/10979df34f3641b088a0cabbc604f67f : SUCCESS in 8m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a679ea323aad47fe8dd13d2d2b5d53bb : SUCCESS in 7m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/69e3a49d2eb34725af109038d92d13db : SUCCESS in 8m 16s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/325482808f5940f4aa58d08c57bea22e : SUCCESS in 1h 07m 13s","accounts_in_message":[],"_revision_number":4},{"id":"c089cdba72cfdd08401ec7b18c02f6be5f7e6d9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:17:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4cfc98c38e1a98a817ad744747257862cce60a82","revisions":{"8f1d565232ebb39ad633a6ece754477f09a32ce9":{"kind":"REWORK","_number":1,"created":"2020-03-13 22:03:46.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/69/713069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/69/713069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/1"}}},"commit":{"parents":[{"commit":"f8776cfc9ef9a72be3b23c10e6afc47e1b4862fb","subject":"Expand retry behavior to cover other python-ovs methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f8776cfc9ef9a72be3b23c10e6afc47e1b4862fb"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-05-21 20:12:12.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-13 22:03:38.000000000","tz":-240},"subject":"Make Event logging more useful by default","message":"Make Event logging more useful by default\n\nLog messages like:\n\nChassisGatewayAgentEvent : Matched Chassis, create, None None\n\nare not particularly useful. This adds a default repr to RowEvents\nand also a Row stringifier (which eventually could go in the ovs\nlib). Subclasses of RowEvent that rely on custom match functions\nshould write their own __repr__ methods.\n\nChange-Id: Ice22858df465b37b973718be718673aab46193ce\n(cherry picked from commit ceea8a23f1e113d46da8b62ecf1cb77acbee24a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8f1d565232ebb39ad633a6ece754477f09a32ce9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8f1d565232ebb39ad633a6ece754477f09a32ce9"}]},"branch":"refs/heads/stable/queens"},"bbe5661cd5bba2383d8faed907e6baba8726ede2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-03 09:10:57.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/69/713069/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/69/713069/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/2"}}},"commit":{"parents":[{"commit":"89bdc6f4f0c01b9f34d62e431e42e561f15c99da","subject":"Unbreak mock-releated autospec issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/89bdc6f4f0c01b9f34d62e431e42e561f15c99da"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-05-21 20:12:12.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-04-03 09:10:57.000000000","tz":0},"subject":"Make Event logging more useful by default","message":"Make Event logging more useful by default\n\nLog messages like:\n\nChassisGatewayAgentEvent : Matched Chassis, create, None None\n\nare not particularly useful. This adds a default repr to RowEvents\nand also a Row stringifier (which eventually could go in the ovs\nlib). Subclasses of RowEvent that rely on custom match functions\nshould write their own __repr__ methods.\n\nChange-Id: Ice22858df465b37b973718be718673aab46193ce\n(cherry picked from commit ceea8a23f1e113d46da8b62ecf1cb77acbee24a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bbe5661cd5bba2383d8faed907e6baba8726ede2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bbe5661cd5bba2383d8faed907e6baba8726ede2"}]},"branch":"refs/heads/stable/queens"},"4c1ea4dc5982552d288506d16103d5fc3e2581dc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-03 20:30:13.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/69/713069/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/69/713069/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/3"}}},"commit":{"parents":[{"commit":"6d56a6ae8855eca3c359b364685614059fa30c66","subject":"update pylint to 1.9.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6d56a6ae8855eca3c359b364685614059fa30c66"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-05-21 20:12:12.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-04-03 20:30:13.000000000","tz":0},"subject":"Make Event logging more useful by default","message":"Make Event logging more useful by default\n\nLog messages like:\n\nChassisGatewayAgentEvent : Matched Chassis, create, None None\n\nare not particularly useful. This adds a default repr to RowEvents\nand also a Row stringifier (which eventually could go in the ovs\nlib). Subclasses of RowEvent that rely on custom match functions\nshould write their own __repr__ methods.\n\nChange-Id: Ice22858df465b37b973718be718673aab46193ce\n(cherry picked from commit ceea8a23f1e113d46da8b62ecf1cb77acbee24a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4c1ea4dc5982552d288506d16103d5fc3e2581dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4c1ea4dc5982552d288506d16103d5fc3e2581dc"}]},"branch":"refs/heads/stable/queens"},"4cfc98c38e1a98a817ad744747257862cce60a82":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2020-04-07 18:37:37.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/69/713069/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/69/713069/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/69/713069/4"}}},"commit":{"parents":[{"commit":"460abdbef808c3d0c4bf9e8f8f33f430cf52df58","subject":"Unbreak mock-releated autospec issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/460abdbef808c3d0c4bf9e8f8f33f430cf52df58"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2019-05-21 20:12:12.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-04-07 18:37:29.000000000","tz":-240},"subject":"Make Event logging more useful by default","message":"Make Event logging more useful by default\n\nLog messages like:\n\nChassisGatewayAgentEvent : Matched Chassis, create, None None\n\nare not particularly useful. This adds a default repr to RowEvents\nand also a Row stringifier (which eventually could go in the ovs\nlib). Subclasses of RowEvent that rely on custom match functions\nshould write their own __repr__ methods.\n\nDepends-On: https://review.opendev.org/#/c/717264/\nChange-Id: Ice22858df465b37b973718be718673aab46193ce\n(cherry picked from commit ceea8a23f1e113d46da8b62ecf1cb77acbee24a4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4cfc98c38e1a98a817ad744747257862cce60a82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/4cfc98c38e1a98a817ad744747257862cce60a82"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
