)]}'
{"id":"openstack%2Fovsdbapp~713072","triplet_id":"openstack%2Fovsdbapp~stable%2Fstein~I7bd306b82973fb4b43fd93d3b6b5886e8dc65418","project":"openstack/ovsdbapp","branch":"stable/stein","hashtags":[],"change_id":"I7bd306b82973fb4b43fd93d3b6b5886e8dc65418","subject":"DNM: Testing functional on stable/stein. Please ignore.","status":"ABANDONED","created":"2020-03-13 22:12:12.000000000","updated":"2020-04-03 09:07:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"59df996914a79a34302bd052bb2e9b6196a9a0c9","_number":713072,"virtual_id_number":713072,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-14 00:46:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":-1,"date":"2020-03-13 22:12:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-14 00:46:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"809cfe80dd9a03bb36cbf0d7f7246dcb2038c659","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-13 22:12:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59e85f4c02d4a869cb2e3099252984b98481d018","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-13 22:12:35.000000000","message":"Patch Set 1: Workflow-1\n\nplease ignore","accounts_in_message":[],"_revision_number":1},{"id":"df80e5b6404b045c90caba47a0940b161c66d64a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 00:46:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/396ddd180ffc49e68f492d30e691fd5f : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2ec3eae45b74c42b5b5da5d2b9da05f : SUCCESS in 3m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e3e65559fac4497ea4d5c22ab3bc97c0 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72e8ce87e1aa4ed8b951062388d45781 : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dba5480bf75a43508d1b7e401c589318 : SUCCESS in 4m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/aee9b57852734e7eaf5a7476b1c93bb3 : FAILURE in 5m 44s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/018deee5d1124b84a0c9e673c08f2976 : SUCCESS in 2h 25m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5cdca439ac2dce863228fab306a2be9d4dd942c1","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-20 11:08:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"35cfe10dea903d9eddc147900ea938b7105443fd","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-01 09:54:49.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"8d1d4697e025dda71befc3f9723addce02d35f07","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-01 09:55:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7592af1d1d1e60aeea2847785cd2cf6e9da357c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 12:15:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/983ab7a1db1949aeaf5cb2b0cae98f68 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/34e2fa1f7c304ddf826456b4a00dccfa : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8397ef4e322246069720a5c2fa902b22 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/245ef33201df497bb67578d6184f88d1 : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9d44a73f98e4f1f9c8f2525541c8135 : SUCCESS in 6m 24s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/89582500edb442569dc2607388276df3 : FAILURE in 7m 12s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/0f95234f466f4a708e8493ae3d28aae9 : SUCCESS in 2h 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"59df996914a79a34302bd052bb2e9b6196a9a0c9","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 09:07:54.000000000","message":"Abandoned\n\nIssue fixed in https://review.opendev.org/#/c/717142/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"13e7124898f420d275684f8ce77b768a350f090f","revisions":{"13e7124898f420d275684f8ce77b768a350f090f":{"kind":"REWORK","_number":1,"created":"2020-03-13 22:12:12.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/72/713072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/72/713072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/72/713072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/72/713072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/72/713072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/72/713072/1"}}},"commit":{"parents":[{"commit":"07b3b48fa42727a99e57f227b9b42cb3044e54f8","subject":"Fix functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/07b3b48fa42727a99e57f227b9b42cb3044e54f8"}]}],"author":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-13 22:12:05.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-13 22:12:05.000000000","tz":-240},"subject":"DNM: Testing functional on stable/stein. Please ignore.","message":"DNM: Testing functional on stable/stein. Please ignore.\n\nChange-Id: I7bd306b82973fb4b43fd93d3b6b5886e8dc65418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/13e7124898f420d275684f8ce77b768a350f090f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/13e7124898f420d275684f8ce77b768a350f090f"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
