)]}'
{"id":"openstack%2Fovsdbapp~713675","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~I59409f8ec1d8a59f94818abd3e3065c824df7b93","project":"openstack/ovsdbapp","branch":"stable/train","hashtags":[],"change_id":"I59409f8ec1d8a59f94818abd3e3065c824df7b93","subject":"WIP Remove eventhandler thread","status":"ABANDONED","created":"2020-03-18 15:19:21.000000000","updated":"2020-03-23 19:16:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fc0714a7ef54c0b9e61969660d048495060cc62d","_number":713675,"virtual_id_number":713675,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-18 18:31:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":-1,"date":"2020-03-18 15:19:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-18 15:19:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 15:19:21.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-03-18 18:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cfdf29478d5c139fc5e1918a78053ffba0e003c4","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-18 15:19:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7629a02ad5e003744be3fe10bab2f8dc52a5f35b","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-18 15:19:39.000000000","message":"Patch Set 1: Workflow-1\n\nDNM","accounts_in_message":[],"_revision_number":1},{"id":"373ceb01bf986afc3c8802a727f707321b8fa6fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 18:31:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d7d3dfcab54522b761a9e06051984f : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59cbe58ff7804a3da0e7c18b1675bece : FAILURE in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c52ba6b1214a407b9e2f3c1109f1680b : FAILURE in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a49b0ef7ad84bdfa9c750b9faf56aee : FAILURE in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be0068898cfc4000b5bd8f75fc7cb569 : FAILURE in 5m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/86796b3e567d4293982e62dd469beb43 : SUCCESS in 7m 08s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/edfaba06560c4e46a4fcca3126f53662 : TIMED_OUT in 3h 05m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fc0714a7ef54c0b9e61969660d048495060cc62d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-03-23 19:16:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d65001f4a3ff2bbf7ac74fc994acd49a454e764d","revisions":{"d65001f4a3ff2bbf7ac74fc994acd49a454e764d":{"kind":"REWORK","_number":1,"created":"2020-03-18 15:19:21.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/75/713675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/75/713675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/75/713675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/75/713675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/75/713675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/75/713675/1"}}},"commit":{"parents":[{"commit":"df707485c1176b04b2056029e0180eada4f045ab","subject":"Log invalid address values in lsp_set_addresses","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/df707485c1176b04b2056029e0180eada4f045ab"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-03-18 14:13:24.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-03-18 15:19:04.000000000","tz":-240},"subject":"WIP Remove eventhandler thread","message":"WIP Remove eventhandler thread\n\nFor now, just looking at how tests react to this.\n\nWe call Idl.run() from the Connection thread. Idl.run() asserts\n\"not self.txn\" and within that run() call is where notify() is\ncalled. RowEventHandler\u0027s notify is called which queues the event\nto be processed in it\u0027s own processing thread. This means that\nwhen we get around to calling the event\u0027s run() in the other thread,\nwe could be processing a transaction, which could change what values\nwe access on the Row objects.\n\nThis removes the RowEventHandler thread which means that\nnotifications are handled in the Connection thread. Since some\nevents themselves create new transactions in their run() methods,\nwe can\u0027t run those inside Idl.run() without making Idl.run()\nrecursive and causing issues. So the notification is still queued\nlike it was in the previous version, and after we call Idl.run(),\nwe call RowEventHandler\u0027s notify_run() to dequeue the notification\nand process it.\n\nChange-Id: I59409f8ec1d8a59f94818abd3e3065c824df7b93\n(cherry picked from commit 1f5e751f441143f1486c2bc1b7d4b38e10938bc0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d65001f4a3ff2bbf7ac74fc994acd49a454e764d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d65001f4a3ff2bbf7ac74fc994acd49a454e764d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
