)]}'
{"id":"openstack%2Fovsdbapp~717103","triplet_id":"openstack%2Fovsdbapp~master~I0cda6044ee7652350369ff4d5d2b0ab8289c127e","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"I0cda6044ee7652350369ff4d5d2b0ab8289c127e","subject":"Don\u0027t let an Exception in one event affect others","status":"MERGED","created":"2020-04-02 19:03:26.000000000","updated":"2020-04-08 20:26:55.000000000","submitted":"2020-04-08 20:17:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717103-1586377042477-ed4f48fb","meta_rev_id":"405ab47d06f0e9b333790a8a06a6d8aa78b21177","_number":717103,"virtual_id_number":717103,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 20:17:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"recommended":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"all":[{"value":1,"date":"2020-04-06 07:42:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2020-04-06 06:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-08 17:21:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-06 06:23:45.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-04-06 07:42:28.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-04-08 20:17:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ccd4c47f90b42664ece595a3c48ad8f5b7fce33c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-02 19:03:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71ecee3c2e2f36ff8606a24b553cc5b029c60c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 21:32:18.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87a99f98d16049b89af828f614c047bb : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6377ddbfffcf451e9ceaa6be897d743a : FAILURE in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55079af18d8f4096ad67b2ed5771f755 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cf1480eefaf451da96bf49f9a92264d : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecd95ddc25684636b4341f810e0f33fa : FAILURE in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4bc6aaf11d4e13969551fc21bccc1c : SUCCESS in 7m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e99e9d9e5c4c411aaf6a15ccb9458547 : SUCCESS in 6m 06s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/1d2638b01f8544a781a6b25e34c9d03d : SUCCESS in 2h 25m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b5443cc9e20ed27f03d79db4e80623b41ef4fe57","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-02 21:51:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed86e101d6a49e6a548c67c2988bd258d5d95d96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 00:43:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8ed3f29cb854d5e98db413914981810 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4e0bd1fd10f48c0ae5f15700ea10c52 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c01c51f0935c46d6b6c949061fbfeadb : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c3780b6dd4240e286efc43cecf5100b : SUCCESS in 10m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0c63759a04849b7a2bf689d4b127179 : FAILURE in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f16c56872b34aa49b44a978ebf241b9 : SUCCESS in 5m 56s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d33b71150728413fa1d346422c02033b : SUCCESS in 7m 55s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/db7af2e5e4f047a2818d0d832ccf1d48 : SUCCESS in 2h 40m 58s","accounts_in_message":[],"_revision_number":2},{"id":"2fbccfabb651697687caa64b299c068742c17069","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-04-03 15:10:16.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6ce5355d8b3fb57bf4995682f6fe35e2672cdfb","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-03 17:43:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f1a905350ec02cd52a1a453380e5081370378e4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-03 17:50:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c55f8f80a2b4deadaa327ae64f97d1fcf4208b73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 20:14:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/550246be505441459c74201301022f7b : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4776cb353267472a9f65926ea9125909 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/213b3120a3c84a548aadcebd23b34287 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84566aa75be8433da022db7c4dd14aa1 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb61553dbe124f1cb400eaa41d89ee81 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f1ff08096a44da9a7810138495529b2 : SUCCESS in 7m 59s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4586a3abd40345fc9ded0be901b12491 : SUCCESS in 6m 27s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/ee5e2fede6444c94a99c4f683145049e : SUCCESS in 2h 09m 22s","accounts_in_message":[],"_revision_number":3},{"id":"849533b906e98da9f61a68f70f55ea088ce1b2cc","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-04-06 06:23:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4c4744af47799cb7517b057e1561c5d024f1420c","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-04-06 07:42:28.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nLGTM, Thanks!","accounts_in_message":[],"_revision_number":3},{"id":"3471baa37a9d6f5e85b3120543089147223d58c2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-08 17:21:12.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"855f3a8d91193545d9550aba8777e02ca3276119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:21:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d612c1711aa5433ab541305a2667b9d35b8c05d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:17:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1c17ed91b959ece1895f489911de5d775d615c4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:17:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c13a64295f8d4777b3a2dd4b820965aa : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a43a2d31772b49dc94abeca0f74c4ff6 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad1fcb89d6e24808b8f5731e8006e987 : SUCCESS in 7m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dcc13e0baeb14c52a9a419ffa7e9c323 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f34de785cc94f41841ef0d5876b0884 : SUCCESS in 9m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f3d9fcf7c8aa455088c1931548758f02 : SUCCESS in 9m 11s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/3588eb50d13045eab059834d0b6b61b1 : SUCCESS in 2h 35m 49s\n\nWarning:\n  Comments left for invalid file ovsdbapp/backend/ovs_idl/idlutils.py","accounts_in_message":[],"_revision_number":3},{"id":"405ab47d06f0e9b333790a8a06a6d8aa78b21177","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 20:26:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0e6918718f4a889de9211a48e6b928 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"76cc8509bca7574cf6185b5ad91903f3eb932e5d","revisions":{"e8133f48681a34f21ae213cc4383d3271555b800":{"kind":"REWORK","_number":1,"created":"2020-04-02 19:03:26.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/03/717103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/03/717103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/1"}}},"commit":{"parents":[{"commit":"85d632327b0daad854151289c190dee0dcc680d5","subject":"Merge \"Do not try to execute transaction if command list is empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/85d632327b0daad854151289c190dee0dcc680d5"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-02 18:59:07.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-02 18:59:07.000000000","tz":-300},"subject":"Don\u0027t let an Exception in one event affect others","message":"Don\u0027t let an Exception in one event affect others\n\nA RowEvent that is added that raises an exception from matches()\nwould cause the loop to abort that checks whether that event\nmatches successive RowEvents. Since it is fairly common in UPDATE\nevents to want to check old.column !\u003d row.column etc. to test\nwhether a RowEvent should fire and \u0027old\u0027 only contains columns that\nhave been modified, it seems reasonable to treat Exceptions as\n\"not matching\" and to debug log enough info in case there was an\nunintended error.\n\nChange-Id: I0cda6044ee7652350369ff4d5d2b0ab8289c127e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e8133f48681a34f21ae213cc4383d3271555b800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e8133f48681a34f21ae213cc4383d3271555b800"}]},"branch":"refs/heads/master"},"33f9231da73e165f2eee234d56538113c1be52b2":{"kind":"REWORK","_number":2,"created":"2020-04-02 21:51:37.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/03/717103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/03/717103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/2"}}},"commit":{"parents":[{"commit":"85d632327b0daad854151289c190dee0dcc680d5","subject":"Merge \"Do not try to execute transaction if command list is empty\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/85d632327b0daad854151289c190dee0dcc680d5"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-02 18:59:07.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-02 21:51:33.000000000","tz":-300},"subject":"Don\u0027t let an Exception in one event affect others","message":"Don\u0027t let an Exception in one event affect others\n\nA RowEvent that is added that raises an exception from matches()\nwould cause the loop to abort that checks whether that event\nmatches successive RowEvents. Since it is fairly common in UPDATE\nevents to want to check old.column !\u003d row.column etc. to test\nwhether a RowEvent should fire and \u0027old\u0027 only contains columns that\nhave been modified, it seems reasonable to treat Exceptions as\n\"not matching\" and to debug log enough info in case there was an\nunintended error.\n\nChange-Id: I0cda6044ee7652350369ff4d5d2b0ab8289c127e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/33f9231da73e165f2eee234d56538113c1be52b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/33f9231da73e165f2eee234d56538113c1be52b2"}]},"branch":"refs/heads/master"},"76cc8509bca7574cf6185b5ad91903f3eb932e5d":{"kind":"REWORK","_number":3,"created":"2020-04-03 17:50:43.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/03/717103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/03/717103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/03/717103/3"}}},"commit":{"parents":[{"commit":"19d11f66aa5b9ccae9eb01a0e395f12fc613acf7","subject":"Merge \"Remove the dependency on the \"mock\" package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/19d11f66aa5b9ccae9eb01a0e395f12fc613acf7"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-02 18:59:07.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-03 17:50:36.000000000","tz":-300},"subject":"Don\u0027t let an Exception in one event affect others","message":"Don\u0027t let an Exception in one event affect others\n\nA RowEvent that is added that raises an exception from matches()\nwould cause the loop to abort that checks whether that event\nmatches successive RowEvents. Since it is fairly common in UPDATE\nevents to want to check old.column !\u003d row.column etc. to test\nwhether a RowEvent should fire and \u0027old\u0027 only contains columns that\nhave been modified, it seems reasonable to treat Exceptions as\n\"not matching\" and to debug log enough info in case there was an\nunintended error.\n\nChange-Id: I0cda6044ee7652350369ff4d5d2b0ab8289c127e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/76cc8509bca7574cf6185b5ad91903f3eb932e5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/76cc8509bca7574cf6185b5ad91903f3eb932e5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
