)]}'
{"id":"openstack%2Fovsdbapp~717142","triplet_id":"openstack%2Fovsdbapp~stable%2Fstein~I508bc07754f11165fb50f5f3f412f6191d27f8b4","project":"openstack/ovsdbapp","branch":"stable/stein","hashtags":[],"change_id":"I508bc07754f11165fb50f5f3f412f6191d27f8b4","subject":"Unbreak mock-releated autospec issue","status":"MERGED","created":"2020-04-02 23:00:20.000000000","updated":"2020-04-03 15:45:58.000000000","submitted":"2020-04-03 15:45:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717142-1585928758464-e8a3316c","meta_rev_id":"53ecc9fc20b5e1bdbcae25ec064bcf97225d0743","_number":717142,"virtual_id_number":717142,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-03 15:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"recommended":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"all":[{"value":1,"date":"2020-04-03 09:02:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-03 06:39:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"date":"2020-04-02 23:06:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-04-03 13:09:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 23:06:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-03 06:39:09.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-04-03 09:02:20.000000000","updated_by":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"reviewer":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"state":"REVIEWER"},{"updated":"2020-04-03 15:45:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9517cf9dcdcb16bd596e95d5125081ea6fa5f437","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-02 23:00:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22c3b783a68930b098316213e59fbc971c4f9ed1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 01:03:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e54d3f41a1d4ed28a3a6692b30095d1 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/98203929376a4ccdb7361ea15b5a3121 : FAILURE in 7m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e2850aeff2194d779b31a81c2e82666d : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/535840fb41df45ecbbbdb901f046b19c : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c5558c7abb24af180a4e085168bdebe : SUCCESS in 7m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dd954920f2244227909a63b1485b59ff : SUCCESS in 7m 40s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/afa1d70aa0d0491f8bc918a456349c59 : SUCCESS in 1h 54m 23s","accounts_in_message":[],"_revision_number":1},{"id":"61924474cf6f68dd300be89a5c04549c56ab014c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-03 01:05:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"77ec082ea29f4855e3deb1eb93f20e7ce6b8589a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 03:11:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6a31296448e470e84925e2a834827de : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6b77d526b7049489120f015480575e8 : SUCCESS in 3m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8491bd967968440fa660d15867785269 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85458ce6ffd34102ba367a3bde29920a : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a365934e4cf45df82ae08d9200fb4f5 : SUCCESS in 5m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dfd55203d04c4c899ccbf4d367aacdce : SUCCESS in 4m 52s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/b3cfa63fbc624b0b99be2c726ecc02bb : SUCCESS in 2h 05m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e8355c0a450bca833b12c3e71afa4ae1f014d09a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-04-03 06:39:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4737a39621bf31cbd5a0aacce11a5c6eef99fe07","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 09:02:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c6235c4786657adf68b2070f57c4d8e94c8f5fe","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 09:03:01.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 45057e3fb1e9c1cf3c66382e7cf21c5ac5f2b136","accounts_in_message":[],"_revision_number":1},{"id":"cf565a454f61bdf6cfb826a4d440190143a86120","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 09:10:11.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 89bdc6f4f0c01b9f34d62e431e42e561f15c99da","accounts_in_message":[],"_revision_number":1},{"id":"90c444c6edfcb41fb454f6c7a8a92e9d1d598d6a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-03 13:09:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a32d97e47c244da4d7f9cc713bed4ddaf808b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 13:10:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f93400b1c3d599909cd0af0c42160a66793f6bbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 15:45:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6498d7ce976f4e7d97322ebc3e3f2773 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff3bf56ab5e949a9b10b203f7f894b16 : SUCCESS in 7m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/427fee1da7db4ae4ac8795288a7b1496 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b935875c1bd847a5b7f6a61403d935a5 : SUCCESS in 6m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edfc481f0be94d0d8ef76ead4c4be655 : SUCCESS in 10m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/97934a88082f4b31bbc0bb4a920b27e7 : SUCCESS in 8m 07s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/508dbff87ee748a58e8e1f5a710cd2c4 : SUCCESS in 2h 29m 00s","accounts_in_message":[],"_revision_number":1},{"id":"53ecc9fc20b5e1bdbcae25ec064bcf97225d0743","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 15:45:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b22e1c4e43ca4f095bff8cb228ef83dd1474a403","revisions":{"b22e1c4e43ca4f095bff8cb228ef83dd1474a403":{"kind":"REWORK","_number":1,"created":"2020-04-02 23:00:20.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/42/717142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/42/717142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/42/717142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/42/717142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/42/717142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/42/717142/1"}}},"commit":{"parents":[{"commit":"07b3b48fa42727a99e57f227b9b42cb3044e54f8","subject":"Fix functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/07b3b48fa42727a99e57f227b9b42cb3044e54f8"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-02 22:57:38.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-02 22:57:38.000000000","tz":-300},"subject":"Unbreak mock-releated autospec issue","message":"Unbreak mock-releated autospec issue\n\nFor some reason, in stable/stein and older, the autospec\u003dTrue in\nthis test started failing. Using python3\u0027s unittest.mock works, so\njust to get this working use that if it exists.\n\nChange-Id: I508bc07754f11165fb50f5f3f412f6191d27f8b4\nCloses-bug: #1867634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b22e1c4e43ca4f095bff8cb228ef83dd1474a403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b22e1c4e43ca4f095bff8cb228ef83dd1474a403"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
