)]}'
{"id":"openstack%2Fovsdbapp~717264","triplet_id":"openstack%2Fovsdbapp~stable%2Fqueens~I508bc07754f11165fb50f5f3f412f6191d27f8b4","project":"openstack/ovsdbapp","branch":"stable/queens","hashtags":[],"change_id":"I508bc07754f11165fb50f5f3f412f6191d27f8b4","subject":"Unbreak mock-releated autospec issue","status":"MERGED","created":"2020-04-03 09:10:11.000000000","updated":"2020-04-08 18:27:32.000000000","submitted":"2020-04-08 18:27:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"717264-1586370452549-02fefbb6","meta_rev_id":"ee87623cd09cdfe7f4109492ca72b76d390e6e2b","_number":717264,"virtual_id_number":717264,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 18:27:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":2,"date":"2020-04-08 15:42:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2020-04-08 17:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2020-04-08 17:18:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 15:13:54.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-04-08 15:42:27.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-08 17:18:03.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-08 18:27:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"74803007bd81cf018446de6a4577f3a80bae2214","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 09:10:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"7f669887ba5c72ab73cbc77446c694bc172ad5ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 10:53:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf0ba54fa60f40499a8d66d40ed3f1f4 : FAILURE in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21663a0ac8474eb2b3594b92c2c8acaf : SUCCESS in 6m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/577579dca3c2462da45394c0ca12ca38 : SUCCESS in 6m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/147d3ec5a38b40329c023361663ea1a9 : SUCCESS in 8m 13s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/4a33a80b8dac486aae3257e999c8fdd9 : SUCCESS in 1h 17m 22s\n\nWarning:\n  Comments left for invalid file tools/debug_venv.py","accounts_in_message":[],"_revision_number":1},{"id":"d6552f7a6ae01644c63470599a6c04f4b4e02961","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-04-03 15:13:54.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks like there is another issue unrelated to this patch","accounts_in_message":[],"_revision_number":1},{"id":"33bc3b235b1bf38ec723b0b519334d83dff67a8a","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 18:38:29.000000000","message":"Patch Set 1:\n\n\u003e Looks like there is another issue unrelated to this patch\n\nIndeed. :^)  See: https://review.opendev.org/#/c/717365/ and friends.","accounts_in_message":[],"_revision_number":1},{"id":"450446ec77d4d6ad01bd06337092183e0dd04ec9","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 21:10:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"efd095c192ea90e9f08defa1ea286ac93afb437b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 22:26:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b715a10c2494f66a947b5df0819044d : SUCCESS in 6m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31ee9dbd4411476b9131551a5a3f9c5d : SUCCESS in 6m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fd2a7b69dfbd4ab0bdab3f172945539a : SUCCESS in 5m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d1b7a2820a0f487b97a9f47d700c1e40 : SUCCESS in 6m 57s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/a58a112c8c694780afcb6484fe89c306 : SUCCESS in 1h 06m 18s\n\nWarning:\n  Job openstack-tox-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional/lib/python3.5/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-py35: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py35/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-py35: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py35/lib/python3.5/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py27/lib/python2.7/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.5/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Comments left for invalid file .tox/py35/lib/python3.5/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/pep8/lib/python3.5/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/py35/lib/python3.5/site-packages/pep8.py\n  Comments left for invalid file .tox/functional/lib/python3.5/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/functional/lib/python3.5/site-packages/pep8.py\n  Comments left for invalid file .tox/py27/lib/python2.7/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/pep8/lib/python3.5/site-packages/pep8.py","accounts_in_message":[],"_revision_number":2},{"id":"0e45830f25ed8ae46d540c044c1b34801c3049ba","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-05 23:37:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8079b5f18c6450ef8bff1bae8dbd49ecaeee43ec","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-07 18:26:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5b4d4c3b45be793ec2a9922c44cecdbf4209aac9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 19:41:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc529b6c4b644e9834c897fc0c78d40 : SUCCESS in 6m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0780d03b9d514c2baaeba01d56a8b398 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/782af09d4a6d42149039513ccd991fec : SUCCESS in 4m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2101a9aa2db546e9bc78dbd3e2c46e28 : SUCCESS in 7m 17s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/9354ff88289f495b9cd55f1424711a4c : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":3},{"id":"d24af02c177d32209ec8e786199c28a16bb224ec","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-07 20:03:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9b2af4bdf3c8477b04950229f765bddf0aa78957","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-08 15:42:27.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0103d616d1947db47f20c00e9a2b0a32d1239595","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-08 17:18:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9a8257626f464e3be0fa5451efa51d50a7602c83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:19:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"483cc335eed6ea3b894e6923f6bc58aceefd5cc9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 18:27:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ee87623cd09cdfe7f4109492ca72b76d390e6e2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 18:27:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed579e301a68470ba2bd07ee3ca16cc5 : SUCCESS in 7m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/264ae6b9d2204bc48cd4ad040114e9a4 : SUCCESS in 7m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/37e9d03fcd1e4774a4324ff3e6ca55f1 : SUCCESS in 7m 47s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2df8c5390116468ba88e5dd607459680 : SUCCESS in 8m 48s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/ee42fc2779db42389e3e1c18caa8f48c : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"460abdbef808c3d0c4bf9e8f8f33f430cf52df58","revisions":{"89bdc6f4f0c01b9f34d62e431e42e561f15c99da":{"kind":"REWORK","_number":1,"created":"2020-04-03 09:10:11.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/64/717264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/64/717264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/1"}}},"commit":{"parents":[{"commit":"f8776cfc9ef9a72be3b23c10e6afc47e1b4862fb","subject":"Expand retry behavior to cover other python-ovs methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f8776cfc9ef9a72be3b23c10e6afc47e1b4862fb"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-02 22:57:38.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-04-03 09:10:11.000000000","tz":0},"subject":"Unbreak mock-releated autospec issue","message":"Unbreak mock-releated autospec issue\n\nFor some reason, in stable/stein and older, the autospec\u003dTrue in\nthis test started failing. Using python3\u0027s unittest.mock works, so\njust to get this working use that if it exists.\n\nChange-Id: I508bc07754f11165fb50f5f3f412f6191d27f8b4\nCloses-bug: #1867634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/89bdc6f4f0c01b9f34d62e431e42e561f15c99da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/89bdc6f4f0c01b9f34d62e431e42e561f15c99da"}]},"branch":"refs/heads/stable/queens"},"66eef4c0ebaf1f05970398d31bd09114b527044e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-03 21:10:30.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/64/717264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/64/717264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/2"}}},"commit":{"parents":[{"commit":"f8776cfc9ef9a72be3b23c10e6afc47e1b4862fb","subject":"Expand retry behavior to cover other python-ovs methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f8776cfc9ef9a72be3b23c10e6afc47e1b4862fb"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-02 22:57:38.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-04-03 21:08:50.000000000","tz":-240},"subject":"Unbreak mock-releated autospec issue","message":"Unbreak mock-releated autospec issue\n\nFor some reason, in stable/stein and older, the autospec\u003dTrue in\nthis test started failing. Using python3\u0027s unittest.mock works, so\njust to get this working use that if it exists.\n\nDepends-On: https://review.opendev.org/#/c/717365/\nDepends-On: https://review.opendev.org/#/c/717370/\nChange-Id: I508bc07754f11165fb50f5f3f412f6191d27f8b4\nCloses-bug: #1867634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/66eef4c0ebaf1f05970398d31bd09114b527044e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/66eef4c0ebaf1f05970398d31bd09114b527044e"}]},"branch":"refs/heads/stable/queens"},"460abdbef808c3d0c4bf9e8f8f33f430cf52df58":{"kind":"REWORK","_number":3,"created":"2020-04-07 18:26:53.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/64/717264/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/64/717264/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/64/717264/3"}}},"commit":{"parents":[{"commit":"f8776cfc9ef9a72be3b23c10e6afc47e1b4862fb","subject":"Expand retry behavior to cover other python-ovs methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f8776cfc9ef9a72be3b23c10e6afc47e1b4862fb"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-02 22:57:38.000000000","tz":-300},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-04-07 18:14:41.000000000","tz":-240},"subject":"Unbreak mock-releated autospec issue","message":"Unbreak mock-releated autospec issue\n\nFor some reason, in stable/stein and older, the autospec\u003dTrue in\nthis test started failing. Using python3\u0027s unittest.mock works, so\njust to get this working use that if it exists.\n\nAlso adding basepython to python2 via tox.ini to keep pep8 working\nin stable/queens.\n\nChange-Id: I508bc07754f11165fb50f5f3f412f6191d27f8b4\nCloses-bug: #1867634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/460abdbef808c3d0c4bf9e8f8f33f430cf52df58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/460abdbef808c3d0c4bf9e8f8f33f430cf52df58"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
