)]}'
{"id":"openstack%2Fovsdbapp~717365","triplet_id":"openstack%2Fovsdbapp~stable%2Fqueens~I08bfe8b35fe250ce04f57e6e3ee435d36063a5a9","project":"openstack/ovsdbapp","branch":"stable/queens","hashtags":[],"change_id":"I08bfe8b35fe250ce04f57e6e3ee435d36063a5a9","subject":"Fix python3 compat with debug_venv.py","status":"ABANDONED","created":"2020-04-03 18:26:45.000000000","updated":"2020-04-07 20:04:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e126981859b68c14972585f96ae7a5e362bf8d8","_number":717365,"virtual_id_number":717365,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-03 21:13:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 18:26:45.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-03 19:37:15.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-04-03 19:37:22.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-03 21:13:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f13218a16607fbf20b290573459710fa9f87f642","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 18:26:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"983e556c979762a726aa8a19b9dc4d913a46437e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 18:38:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47bcc3aa6231437f96a514dc89cd7d94 : FAILURE in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ad2ea2314894c218ed6d6eddbf50de9 : SUCCESS in 5m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/07b653246ce84f15b4a12340819de1c6 : SUCCESS in 5m 01s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/326e6cae0072432e88e2275f82467c94 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0607150b2af54944e25129e7514ab5b07170aa85","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 19:39:49.000000000","message":"Patch Set 1:\n\nIn order to fix the ci ussue, this gerri needs https://review.opendev.org/#/c/717370/","accounts_in_message":[],"_revision_number":1},{"id":"2de1ce6f17e23257fc7cf0cd7b96bdcde62cf8ca","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 21:01:08.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cbc11b3a977c463ba2a0bf0f6ceede643b71b60f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 21:13:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe458df68434396b99292669b983eba : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd0be53cbf51441bb61754638c16ddc1 : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/389457061e3444bca63b7f941e0f0529 : SUCCESS in 8m 13s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9baed60d2c074dc1aea3d77fb0b1c6b1 : SUCCESS in 6m 48s\n\nWarning:\n  Job openstack-tox-py27: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py27/lib/python2.7/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python3.5/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-py35: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py35/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-py35: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py35/lib/python3.5/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional/lib/python3.5/site-packages/setuptools/dist.py in the commit\u0027\n  Job openstack-tox-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Comments left for invalid file .tox/py35/lib/python3.5/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/pep8/lib/python3.5/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/py35/lib/python3.5/site-packages/pep8.py\n  Comments left for invalid file .tox/functional/lib/python3.5/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/functional/lib/python3.5/site-packages/pep8.py\n  Comments left for invalid file .tox/py27/lib/python2.7/site-packages/setuptools/dist.py\n  Comments left for invalid file .tox/pep8/lib/python3.5/site-packages/pep8.py","accounts_in_message":[],"_revision_number":2},{"id":"2e126981859b68c14972585f96ae7a5e362bf8d8","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-07 20:04:30.000000000","message":"Abandoned\n\nChanging pylint versions on Queens is not a good idea. Thanks to bhaley, we found a simpler solution to making Queens remain working as a python 2 only:  https://review.opendev.org/#/c/717264/2..3   Abandon this ship!","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2256128a1d1bb2cf9c88c782be468449abeb77e","revisions":{"8a758cd08b88d019a09c3f65604b76e5ed884000":{"kind":"REWORK","_number":1,"created":"2020-04-03 18:26:45.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/65/717365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/65/717365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/65/717365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/65/717365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/65/717365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/65/717365/1"}}},"commit":{"parents":[{"commit":"89bdc6f4f0c01b9f34d62e431e42e561f15c99da","subject":"Unbreak mock-releated autospec issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/89bdc6f4f0c01b9f34d62e431e42e561f15c99da"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2018-07-02 12:33:22.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-04-03 18:26:32.000000000","tz":-240},"subject":"Fix python3 compat with debug_venv.py","message":"Fix python3 compat with debug_venv.py\n\nChange-Id: I08bfe8b35fe250ce04f57e6e3ee435d36063a5a9\n(cherry picked from commit 2ee22a5484093221d76b2bc1c3c540ba15008a49)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8a758cd08b88d019a09c3f65604b76e5ed884000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8a758cd08b88d019a09c3f65604b76e5ed884000"}]},"branch":"refs/heads/stable/queens"},"a2256128a1d1bb2cf9c88c782be468449abeb77e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-03 21:01:08.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/65/717365/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/65/717365/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/65/717365/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/65/717365/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/65/717365/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/65/717365/2"}}},"commit":{"parents":[{"commit":"89bdc6f4f0c01b9f34d62e431e42e561f15c99da","subject":"Unbreak mock-releated autospec issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/89bdc6f4f0c01b9f34d62e431e42e561f15c99da"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2018-07-02 12:33:22.000000000","tz":0},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-04-03 20:58:26.000000000","tz":-240},"subject":"Fix python3 compat with debug_venv.py","message":"Fix python3 compat with debug_venv.py\n\nDepends-On: https://review.opendev.org/#/c/717370/\nChange-Id: I08bfe8b35fe250ce04f57e6e3ee435d36063a5a9\n(cherry picked from commit 2ee22a5484093221d76b2bc1c3c540ba15008a49)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a2256128a1d1bb2cf9c88c782be468449abeb77e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a2256128a1d1bb2cf9c88c782be468449abeb77e"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
