)]}'
{"id":"openstack%2Fovsdbapp~717370","triplet_id":"openstack%2Fovsdbapp~stable%2Fqueens~I91a304241ebdb4d6bd15509cae8463d7b43b3c0f","project":"openstack/ovsdbapp","branch":"stable/queens","hashtags":[],"change_id":"I91a304241ebdb4d6bd15509cae8463d7b43b3c0f","subject":"update pylint to 1.9.2","status":"ABANDONED","created":"2020-04-03 18:55:41.000000000","updated":"2020-04-07 20:03:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76c107f6fd035aed638bf967c209050f82537d7e","_number":717370,"virtual_id_number":717370,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-07 15:36:37.000000000","_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-04-03 20:20:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"value":0,"date":"2020-04-06 20:56:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-03 19:38:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"date":"2020-04-03 19:38:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 18:55:41.000000000","updated_by":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"reviewer":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"state":"REVIEWER"},{"updated":"2020-04-03 19:38:06.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-04-03 19:38:13.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-03 20:20:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-06 20:56:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"c3b9e80474c64dbb240548d3c2fc906aa73f165c","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 18:55:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a450cfe4b6c679cc273912cd026bd765c8053afd","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 19:15:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"979c283a8221335d6b03b6cade09f301391fda92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 20:20:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ea2995cbdc45d6bcdd3cb43e09e73d : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea99b2aae6f0436ebed5e84cac2d82d0 : SUCCESS in 5m 41s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fdc2b6c2605943c89b2fa6397de02e03 : SUCCESS in 4m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/205323876aca44f58f7f735109beba3a : SUCCESS in 4m 44s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4eddf96b63b340328f4f79dbf95d4b69 : SUCCESS in 6m 52s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/17c15e2882f545f6bf66c3704674cd22 : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":2},{"id":"459f5777060490e582976daa7168e6ad99618c61","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-03 20:30:48.000000000","message":"Patch Set 2: Code-Review+2\n\nSince this change merged in stein, does it need to go to stable/rocky as well?","accounts_in_message":[],"_revision_number":2},{"id":"1e2891248a73b9578f1d06c210f4e5002739519e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 20:42:15.000000000","message":"Patch Set 2:\n\n\u003e Since this change merged in stein, does it need to go to\n \u003e stable/rocky as well?\n\nhm... good point. Created: https://review.opendev.org/717387  for that.","accounts_in_message":[],"_revision_number":2},{"id":"61b08ae69efb249cbcb61532b9a75749b4966ef8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-04-06 20:56:20.000000000","message":"Patch Set 2: -Code-Review\n\nI\u0027m not having second thoughts on this, since it bumps the test requirements of stable/stein which could break something.  Can we just force pep8 to run under py2?","accounts_in_message":[],"_revision_number":2},{"id":"dea16d65385d17f8c4fd0c1089135619274ef5dc","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-07 15:36:37.000000000","message":"Patch Set 2:\n\n\u003e I\u0027m not having second thoughts on this, since it bumps the test\n \u003e requirements of stable/stein which could break something.  Can we\n \u003e just force pep8 to run under py2?\n\nI thinks that is the crux of the problem. Right now, the upstream ci for queens is broken and this makes it difficult for us to merge anything because the failure will be unrelated to what we are trying to address.\n\nI\u0027m sure this used to work at one point, but then changes exgternal to this repo broke ci. What do you recommend?","accounts_in_message":[],"_revision_number":2},{"id":"76c107f6fd035aed638bf967c209050f82537d7e","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-07 20:03:46.000000000","message":"Abandoned\n\nChanging pylint versions on Queens is not a good idea. Thanks to bhaley, we found a simpler solution to making Queens remain working as a python 2 only:  https://review.opendev.org/#/c/717264/2..3   Abandon this ship!","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d56a6ae8855eca3c359b364685614059fa30c66","revisions":{"d1bb6684eb61554dcb9ef9f7c6bad08f718146cd":{"kind":"REWORK","_number":1,"created":"2020-04-03 18:55:41.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/717370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/717370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/717370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/717370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/717370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/717370/1"}}},"commit":{"parents":[{"commit":"8a758cd08b88d019a09c3f65604b76e5ed884000","subject":"Fix python3 compat with debug_venv.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8a758cd08b88d019a09c3f65604b76e5ed884000"}]}],"author":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-08-21 13:25:29.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-04-03 18:54:56.000000000","tz":-240},"subject":"update pylint to 1.9.2","message":"update pylint to 1.9.2\n\nThe older version of pylint being used does not work\ncorrectly under python 3. In order to be able to\nupdate the pylint job to run under python 3,\nwe need to update the tool.\n\nAslo fix a couple of pylint issues:\n\n- Fix import order issues.\n- Fix \u0027no-else-return\u0027 checks.\n\nAdded the following to the blacklist:\n\n- inconsistent-return-statements\n- catching-non-exception\n- using-constant-test\n\nChange-Id: I91a304241ebdb4d6bd15509cae8463d7b43b3c0f\nSigned-off-by: Chuck Short \u003cchucks@redhat.com\u003e\n(cherry picked from commit ac3c95c1b4c33197812fd7df43a6e71bf0251475)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d1bb6684eb61554dcb9ef9f7c6bad08f718146cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d1bb6684eb61554dcb9ef9f7c6bad08f718146cd"}]},"branch":"refs/heads/stable/queens"},"6d56a6ae8855eca3c359b364685614059fa30c66":{"kind":"REWORK","_number":2,"created":"2020-04-03 19:15:17.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/70/717370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/70/717370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/717370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/717370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/70/717370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/70/717370/2"}}},"commit":{"parents":[{"commit":"8a758cd08b88d019a09c3f65604b76e5ed884000","subject":"Fix python3 compat with debug_venv.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8a758cd08b88d019a09c3f65604b76e5ed884000"}]}],"author":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-08-21 13:25:29.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-04-03 19:13:05.000000000","tz":-240},"subject":"update pylint to 1.9.2","message":"update pylint to 1.9.2\n\nThe older version of pylint being used does not work\ncorrectly under python 3. In order to be able to\nupdate the pylint job to run under python 3,\nwe need to update the tool.\n\nAslo fix a couple of pylint issues:\n\n- Fix import order issues.\n- Fix \u0027no-else-return\u0027 checks.\n- Undefined variable \u0027unicode\u0027 in ovsdbapp.exceptions\n  (ref commit: 665666c031a577955642e96d7b3491336997cfa9)\n\nAdded the following to the blacklist:\n\n- inconsistent-return-statements\n- catching-non-exception\n- using-constant-test\n\nChange-Id: I91a304241ebdb4d6bd15509cae8463d7b43b3c0f\nSigned-off-by: Chuck Short \u003cchucks@redhat.com\u003e\n(cherry picked from commit ac3c95c1b4c33197812fd7df43a6e71bf0251475)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6d56a6ae8855eca3c359b364685614059fa30c66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6d56a6ae8855eca3c359b364685614059fa30c66"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
