)]}'
{"id":"openstack%2Fovsdbapp~717387","triplet_id":"openstack%2Fovsdbapp~stable%2Frocky~I91a304241ebdb4d6bd15509cae8463d7b43b3c0f","project":"openstack/ovsdbapp","branch":"stable/rocky","hashtags":[],"change_id":"I91a304241ebdb4d6bd15509cae8463d7b43b3c0f","subject":"update pylint to 1.9.2","status":"ABANDONED","created":"2020-04-03 20:41:45.000000000","updated":"2020-04-06 15:12:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1545075caa4c250e77bbff19d3fdccba5f9c5660","_number":717387,"virtual_id_number":717387,"owner":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-03 21:32:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-03 20:41:45.000000000","updated_by":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"reviewer":{"_account_id":24,"name":"Chuck Short","email":"charles.short@windriver.com","username":"zulcss"},"state":"REVIEWER"},{"updated":"2020-04-03 20:43:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-04-03 21:32:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5dcb64ee6ff486b698c617cd2cb021a4ba526b5d","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 20:41:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eacc75df26687f03eadf125f27a156855d723411","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-03 20:55:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d9eadf77c725441597b83e51106a75b15870604c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 21:32:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00e3fc26956c4163bea108538cb61713 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0617da85368441039d50ec25c735a7c3 : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f5dd7afafd194427bd0398f0bea6d853 : SUCCESS in 4m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/24a4fadf33e048ec970169a86612fe91 : FAILURE in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64aa3be0753b4b1d85dbe6399c959fb0 : FAILURE in 12m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5f0c33bf866f4f12a15f7d8b87f30d7a : SUCCESS in 7m 08s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/9c86ef13c52c4628b65ec7d4a258fb7e : FAILURE in 34m 33s\n\nWarning:\n  Job openstack-tox-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional/src/ovs/Documentation/conf.py in the commit\u0027\n  Job openstack-tox-functional: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/functional/lib/python3.5/site-packages/pep8.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python2.7/site-packages/isort/settings.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python2.7/site-packages/backports/configparser/__init__.py in the commit\u0027\n  Job openstack-tox-pep8: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/pep8/lib/python2.7/site-packages/isort/settings.py in the commit\u0027\n  Comments left for invalid file .tox/functional/lib/python3.5/site-packages/pep8.py\n  Comments left for invalid file .tox/pep8/lib/python2.7/site-packages/backports/configparser/__init__.py\n  Comments left for invalid file .tox/functional/src/ovs/Documentation/conf.py\n  Comments left for invalid file .tox/pep8/lib/python2.7/site-packages/isort/settings.py","accounts_in_message":[],"_revision_number":2},{"id":"1545075caa4c250e77bbff19d3fdccba5f9c5660","author":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"date":"2020-04-06 15:12:19.000000000","message":"Abandoned\n\nAbandoning support for these changes on stable/rocky.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dce1a53eb71ece676726029a9049cd4e42ef2356","revisions":{"0f3e92443ec783177b25200e8f26ba60fbcd8614":{"kind":"REWORK","_number":1,"created":"2020-04-03 20:41:45.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/87/717387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/87/717387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/717387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/717387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/717387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/87/717387/1"}}},"commit":{"parents":[{"commit":"2b1552d8b044750afab5a0e43b5354b3b1d3e28d","subject":"Expand retry behavior to cover other python-ovs methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2b1552d8b044750afab5a0e43b5354b3b1d3e28d"}]}],"author":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-08-21 13:25:29.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flaviof@redhat.com","date":"2020-04-03 20:41:12.000000000","tz":-240},"subject":"update pylint to 1.9.2","message":"update pylint to 1.9.2\n\nThe older version of pylint being used does not work\ncorrectly under python 3. In order to be able to\nupdate the pylint job to run under python 3,\nwe need to update the tool.\n\nAslo fix a couple of pylint issues:\n\n- Fix import order issues.\n- Fix \u0027no-else-return\u0027 checks.\n\nAdded the following to the blacklist:\n\n- inconsistent-return-statements\n- catching-non-exception\n- using-constant-test\n\nChange-Id: I91a304241ebdb4d6bd15509cae8463d7b43b3c0f\nSigned-off-by: Chuck Short \u003cchucks@redhat.com\u003e\n(cherry picked from commit ac3c95c1b4c33197812fd7df43a6e71bf0251475)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0f3e92443ec783177b25200e8f26ba60fbcd8614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0f3e92443ec783177b25200e8f26ba60fbcd8614"}]},"branch":"refs/heads/stable/rocky"},"dce1a53eb71ece676726029a9049cd4e42ef2356":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-03 20:55:10.000000000","uploader":{"_account_id":11952,"name":"Flavio Fernandes","email":"flavio@flaviof.com","username":"ffernand"},"ref":"refs/changes/87/717387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/87/717387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/717387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/717387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/87/717387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/87/717387/2"}}},"commit":{"parents":[{"commit":"45057e3fb1e9c1cf3c66382e7cf21c5ac5f2b136","subject":"Unbreak mock-releated autospec issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/45057e3fb1e9c1cf3c66382e7cf21c5ac5f2b136"}]}],"author":{"name":"Chuck Short","email":"chucks@redhat.com","date":"2018-08-21 13:25:29.000000000","tz":-240},"committer":{"name":"Flavio Fernandes","email":"flavio@flaviof.com","date":"2020-04-03 20:55:10.000000000","tz":0},"subject":"update pylint to 1.9.2","message":"update pylint to 1.9.2\n\nThe older version of pylint being used does not work\ncorrectly under python 3. In order to be able to\nupdate the pylint job to run under python 3,\nwe need to update the tool.\n\nAslo fix a couple of pylint issues:\n\n- Fix import order issues.\n- Fix \u0027no-else-return\u0027 checks.\n\nAdded the following to the blacklist:\n\n- inconsistent-return-statements\n- catching-non-exception\n- using-constant-test\n\nChange-Id: I91a304241ebdb4d6bd15509cae8463d7b43b3c0f\nSigned-off-by: Chuck Short \u003cchucks@redhat.com\u003e\n(cherry picked from commit ac3c95c1b4c33197812fd7df43a6e71bf0251475)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dce1a53eb71ece676726029a9049cd4e42ef2356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/dce1a53eb71ece676726029a9049cd4e42ef2356"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
