)]}'
{"id":"openstack%2Fovsdbapp~718155","triplet_id":"openstack%2Fovsdbapp~master~I17c24fd88adecbea09c71741c0105df1d0240e66","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"I17c24fd88adecbea09c71741c0105df1d0240e66","subject":"Accept a Row as a result in AddCommand","status":"MERGED","created":"2020-04-07 15:53:49.000000000","updated":"2020-04-09 09:41:51.000000000","submitted":"2020-04-09 09:38:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718155-1586425117872-85e1605a","meta_rev_id":"67af37aaed8a51b02cf0a0cdf6733118a24c3727","_number":718155,"virtual_id_number":718155,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-09 09:38:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2020-04-07 19:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2020-04-09 07:03:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2020-04-09 07:03:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 19:39:25.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-04-09 07:03:12.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-04-09 09:38:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af013e6ede9cb5892d3a58734d8bf07192f5459f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-07 15:53:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0954908ed46aa955150210c05203ac7cdf3213c8","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-04-07 15:56:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e17884dbdaddbab8fe44ecdfce1e1136f975ff3d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-07 16:37:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20c50b4ef031c67d53dacacfa9666ba999fb0fdc","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-07 18:08:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09ce2275ea4f6b78f25e157c659bac6995f2501b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-07 18:15:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5332c1094a02050f9ee70d688da727a2736cad9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 18:33:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ef5ad63a86e4b2cb9c988fc6fc1a566 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc6c2e232594402990ae7edab99b262a : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caf3480c8889408e9df2703db5c94c52 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f1e43b2096848758e1c3eb20be314e1 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/660a125a4be44d589d730ff236b5639e : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f3769a52e014ed59297d8912fa36ab5 : SUCCESS in 8m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f01e7ea6a02b486ca152a5ec45fb45c0 : SUCCESS in 9m 03s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/aa9c1770cec442f7af45ca7921ff5c2f : SUCCESS in 2h 35m 20s\n\nWarning:\n  Comments left for invalid file ovsdbapp/backend/ovs_idl/idlutils.py","accounts_in_message":[],"_revision_number":1},{"id":"ca0217f29949ea106da40499998cf0bcc7c018ec","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-07 18:47:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ecf965d47bf5bc83c7b013afbbd8ca9d434e3480","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-07 18:47:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0e954ca30189b0e3652b75e986eb9b32db8fbe99","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-07 19:39:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5cec586879090d4831ec8e4d3656a593d2ccfc37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 22:20:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cf12f8bfd124ac1bc2737dfa2bc9c4b : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f62d15be5324549a576773f38973e03 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14fd262d218b4503b49a3caf02e317f4 : SUCCESS in 6m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/45cc82cabfa64e12ab132a10040b3455 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ffc970bf454403689d7f9d01853f2c4 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c58ce7ff8541eaa11588acc8ca5325 : SUCCESS in 9m 46s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dd2f4843053f4bdfa7b0e0e9e2df7257 : SUCCESS in 9m 36s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/ab2cb551f1b142c8ac3acfa21989c636 : SUCCESS in 3h 08m 26s\n\nWarning:\n  Comments left for invalid file ovsdbapp/backend/ovs_idl/idlutils.py","accounts_in_message":[],"_revision_number":2},{"id":"3e9c2d43ff45d34df355c147c274ac8b077c2110","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-04-09 07:03:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4bbf535191c175aee1305e8aaa280ce7c26ce410","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 07:03:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"497f43548a51760b9af8c8b4acc578c4e234ec0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:38:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"48d98dee5ee3a5fcd7bf7bc4cb345d92398e877f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:38:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6af00a9e2344f9385c388056d959cb3 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e630d03077342ccb0aed6f9de03fbe0 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c012dd24f6744f8b003b11c1b3b8ba9 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/54f5f7fde1f84f96a49925536adce89f : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e4d1893fa154f9bb0c7631654666a56 : SUCCESS in 7m 54s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f31c23d923824b25b1e649cf64238dea : SUCCESS in 7m 50s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/e384952fa1324ce88cc769f34c4135f8 : SUCCESS in 2h 28m 34s","accounts_in_message":[],"_revision_number":2},{"id":"67af37aaed8a51b02cf0a0cdf6733118a24c3727","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 09:41:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa7494338404ad1898bd809aeae29cb : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0338c0bb5cd6368edc14bff8e64f980d4a6286ac","revisions":{"da4270d2ce208135f90fa33cee4a6cbff1ea0379":{"kind":"REWORK","_number":1,"created":"2020-04-07 15:53:49.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/55/718155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/55/718155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/55/718155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/55/718155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/55/718155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/55/718155/1"}}},"commit":{"parents":[{"commit":"19d11f66aa5b9ccae9eb01a0e395f12fc613acf7","subject":"Merge \"Remove the dependency on the \"mock\" package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/19d11f66aa5b9ccae9eb01a0e395f12fc613acf7"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-07 15:52:20.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-07 15:52:20.000000000","tz":-300},"subject":"Accept a Row as a result in AddCommand","message":"Accept a Row as a result in AddCommand\n\nMake AddCommand a little more lenient and accept Row/RowView objs\nas results.\n\nChange-Id: I17c24fd88adecbea09c71741c0105df1d0240e66\nRelated-Bug: 1871104\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/da4270d2ce208135f90fa33cee4a6cbff1ea0379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/da4270d2ce208135f90fa33cee4a6cbff1ea0379"}]},"branch":"refs/heads/master"},"0338c0bb5cd6368edc14bff8e64f980d4a6286ac":{"kind":"REWORK","_number":2,"created":"2020-04-07 18:47:36.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/55/718155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/55/718155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/55/718155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/55/718155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/55/718155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/55/718155/2"}}},"commit":{"parents":[{"commit":"19d11f66aa5b9ccae9eb01a0e395f12fc613acf7","subject":"Merge \"Remove the dependency on the \"mock\" package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/19d11f66aa5b9ccae9eb01a0e395f12fc613acf7"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-07 15:52:20.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-07 18:47:32.000000000","tz":-300},"subject":"Accept a Row as a result in AddCommand","message":"Accept a Row as a result in AddCommand\n\nMake AddCommand a little more lenient and accept Row/RowView objs\nas results.\n\nChange-Id: I17c24fd88adecbea09c71741c0105df1d0240e66\nRelated-Bug: 1871104\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0338c0bb5cd6368edc14bff8e64f980d4a6286ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0338c0bb5cd6368edc14bff8e64f980d4a6286ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
