)]}'
{"id":"openstack%2Fovsdbapp~720531","triplet_id":"openstack%2Fovsdbapp~master~Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1873305","hashtags":[],"change_id":"Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0","subject":"Allow read-only Commands to bypass txns in execute()","status":"MERGED","created":"2020-04-16 15:57:32.000000000","updated":"2020-04-23 16:19:28.000000000","submitted":"2020-04-23 16:17:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720531-1587658664186-dfe5b5ea","meta_rev_id":"3d542bfd13617eecc5131f3f59d3b85a1ffedb41","_number":720531,"virtual_id_number":720531,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 16:17:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-20 19:00:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2020-04-23 13:14:14.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-23 14:16:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"date":"2020-04-23 13:14:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 13:14:05.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-04-23 13:14:14.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-04-23 14:16:15.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-23 16:17:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3b50149b5209f51a85687bb916f442647a9859f2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-16 15:57:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2616c9fc74314332d8feeb95b8126c1cadf72570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 18:21:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08df97694bf945d4a7f4498c0873819b : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ed3d50df6149c694b3f1e536e2e513 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8deddf7a3854302bfce8393b4b91fc2 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4e34111b1eea415692aaa2941ee0e9b0 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9edfbf19cc4441c78708273f71f2467b : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051ca1929aef42f5a1e6147ca42de6a6 : SUCCESS in 5m 30s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/31d696d483844318978ee61e14dfddb5 : SUCCESS in 6m 56s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/ee9b375d7fc946e384cebb6edf2eb5e1 : FAILURE in 2h 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"67a663216da1d97f4664be9ffef3aeae010e7f0b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-16 18:59:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b34d55077c5524227a161621b897fa1d365d11ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 21:34:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d40b53b0465f4e39badf55b4bc0ab334 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5ca4641ef264f0bb50b1da9c8d7e962 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1435bb9b29ee4b9bad48b83caf5abc60 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01b8816671054023b045b60f8f974ef1 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcd1921a4ea3475f97837ed64a8dbd58 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f59cfb41892843cfbbc2f011fb3f8c33 : SUCCESS in 4m 41s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/60cba8eb04dc4393819ca998c4db1e90 : SUCCESS in 5m 55s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/dc951aac28554bfb94c26fb442844e6c : FAILURE in 2h 26m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d1b6cc82ab146c1073f19881d59ed62812763913","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-17 08:56:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ee42dfe0c4ab5413b29b188b209e343e3349de18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-17 11:42:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4aa79f9a795c41029762fc467c5ff312 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a2d6045ea0c41458c40751f00e6fd14 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f17735db28af496a9a464511af41e87c : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9c4328081984a929ccc97934cff75a5 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed79ccc684b340a686b706131d2e4177 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859538b79f6041b5969d3049e20f33b9 : SUCCESS in 12m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/3f54131a4ac14bf6b30dd7107793d627 : SUCCESS in 7m 05s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/4905581b41cd4c3984050c6b876aa998 : FAILURE in 2h 29m 26s","accounts_in_message":[],"_revision_number":2},{"id":"1c5ee05dda8e48f0bf381fc8acebf624ecd8a94b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-17 21:41:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"96cad1abd9e9a3f279d68282e1f10e385f7465f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 00:30:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/596a08c6f6d74349afe742dc11a206d4 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e75095202c16498ea77dbefb6ee3d64a : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dd32ef8e05e44368a407070ca158678 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e2cbc23c13e45ca908709fd56e4a9b9 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbe4cb137bbe4003a54b189324fe9727 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cc9f1861314f1c81c418c9fc1b5fd8 : SUCCESS in 5m 29s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/c48bf41455364dd6ae9ffd46c279e34e : SUCCESS in 8m 17s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/5bae1dd50d5546a7b81717fbaba4827c : FAILURE in 2h 40m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5250b193ab981396c41fe35b9164fdac612006c9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-19 23:25:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e57676d6b7140a41884fd0e6716d1524b0944330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 01:10:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5c589620b8b4abfa31aaef7a71f2839 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58aa935bc0ed4589861e953dd43b5104 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4f127753b4b4c25987dfd21e9102482 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84fcbe2daeaf42bc8c802821698a2664 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f7a04ecd2a44a60a7792e4738b1fddf : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5058033959ba4f8dbd5bfb54d6013fa1 : SUCCESS in 5m 17s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4fb8bc7beb2b42de97cb06d75a67cab2 : SUCCESS in 5m 41s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e1f75665390146469d54770dc0f4cead : SUCCESS in 1h 41m 37s","accounts_in_message":[],"_revision_number":3},{"id":"bafb02679669fcc67ec80c3483a1bfd3d222c5e0","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-20 19:00:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c950bbb2a164faa702a8a6d9a5ea7f71e324093f","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-23 14:16:15.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"72ef58c3d10687aa8c8a9518b122f88b37772c96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 14:16:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"41cdb6eb36f1c4085ab72d32480165f926f18316","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-23 14:55:17.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 385663fd92a5abcc1836d40b62095c1b5d884fea","accounts_in_message":[],"_revision_number":3},{"id":"93a094440f2e8d59e10c9f015b71d5e18ec508d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:17:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7801b4955734037808be3a1e9988a3e : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/638c3e6b1c6f4a02b63869e2ec09a854 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e33f9491d22e48d68cadf9f07a1bcffb : SUCCESS in 8m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/139b206aefa74c71b4d87abe617e1e9e : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/935c4c86abdc4581b0d2035e5784121d : SUCCESS in 5m 48s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a6815c48a9d745d790a6e0ddfd1c399f : SUCCESS in 8m 02s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/98742d8301e340a2aa19d9c00236416a : SUCCESS in 1h 53m 39s","accounts_in_message":[],"_revision_number":3},{"id":"06ad2a2bc754b370744637a04c3dc59f480eb352","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:17:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"dae4ff5133fbbf10480dd5b1a2c63f77ecc533db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:19:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198728851bb4445991b15e61d4ef4fcf : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"aadf1624ef9f8c8b92601ccb973775f508f7f11c","revisions":{"44a12084128b0e4546faff5758ec36a43005e623":{"kind":"REWORK","_number":1,"created":"2020-04-16 15:57:32.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/720531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/720531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/1"}}},"commit":{"parents":[{"commit":"e8ac095c897f5c2905981061764fddab7d77cd39","subject":"Merge \"Accept a Row as a result in AddCommand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e8ac095c897f5c2905981061764fddab7d77cd39"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-16 15:55:25.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-16 15:57:24.000000000","tz":0},"subject":"[WIP] Allow read-only Commands to bypass txns in execute()","message":"[WIP] Allow read-only Commands to bypass txns in execute()\n\nChange-Id: Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0\nCloses-Bug: #1873305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/44a12084128b0e4546faff5758ec36a43005e623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/44a12084128b0e4546faff5758ec36a43005e623"}]},"branch":"refs/heads/master"},"e935ee52f355fe70c066ff15e9fc7eede9bf99b5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-17 08:56:24.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/720531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/720531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/2"}}},"commit":{"parents":[{"commit":"e8ac095c897f5c2905981061764fddab7d77cd39","subject":"Merge \"Accept a Row as a result in AddCommand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e8ac095c897f5c2905981061764fddab7d77cd39"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-16 15:55:25.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 08:51:22.000000000","tz":0},"subject":"Allow read-only Commands to bypass txns in execute()","message":"Allow read-only Commands to bypass txns in execute()\n\nThis patch continues https://review.opendev.org/#/c/633070/.\n\nSince the ovs_idl backend maintains an in-memory copy of the DB,\nread-only commands can be executed w/o creating a transaction.\nSince it is possible for a function that calls execute() to be\nautomatically nested into an existing transaction, execute()\nwasn\u0027t able to access the result since it only happens once the\nparent transaction completed. Handling these read-only commands\nw/o creating a transaction allows the results to be read and\navoids wasting resources by creating needless transactions.\n\nReadOnlyCommands can still be used in with transaction(): blocks\nwith no changes.\n\nChange-Id: Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0\nCloses-Bug: #1873305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e935ee52f355fe70c066ff15e9fc7eede9bf99b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e935ee52f355fe70c066ff15e9fc7eede9bf99b5"}]},"branch":"refs/heads/master"},"aadf1624ef9f8c8b92601ccb973775f508f7f11c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-19 23:25:41.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/31/720531/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/31/720531/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/31/720531/3"}}},"commit":{"parents":[{"commit":"359863155811ee82fae1cf9f07d0375d3b61c519","subject":"Tempest job inherits from Neutron tempest OVN job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/359863155811ee82fae1cf9f07d0375d3b61c519"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-16 15:55:25.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-19 23:25:41.000000000","tz":0},"subject":"Allow read-only Commands to bypass txns in execute()","message":"Allow read-only Commands to bypass txns in execute()\n\nThis patch continues https://review.opendev.org/#/c/633070/.\n\nSince the ovs_idl backend maintains an in-memory copy of the DB,\nread-only commands can be executed w/o creating a transaction.\nSince it is possible for a function that calls execute() to be\nautomatically nested into an existing transaction, execute()\nwasn\u0027t able to access the result since it only happens once the\nparent transaction completed. Handling these read-only commands\nw/o creating a transaction allows the results to be read and\navoids wasting resources by creating needless transactions.\n\nReadOnlyCommands can still be used in with transaction(): blocks\nwith no changes.\n\nChange-Id: Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0\nCloses-Bug: #1873305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/aadf1624ef9f8c8b92601ccb973775f508f7f11c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/aadf1624ef9f8c8b92601ccb973775f508f7f11c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
