)]}'
{"id":"openstack%2Fovsdbapp~720880","triplet_id":"openstack%2Fovsdbapp~master~I910ae5d4d32b43d7af15f196f4af6f5e1c4c0734","project":"openstack/ovsdbapp","branch":"master","topic":"tempest_inherit_neutron","hashtags":[],"change_id":"I910ae5d4d32b43d7af15f196f4af6f5e1c4c0734","subject":"Tempest job inherits from Neutron tempest OVN job","status":"MERGED","created":"2020-04-18 11:22:45.000000000","updated":"2020-05-19 20:58:59.000000000","submitted":"2020-04-19 18:50:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720880-1587322226238-7130f63b","meta_rev_id":"ca26fab56cc1da1c7d37ed041ee0179198b4dffc","_number":720880,"virtual_id_number":720880,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-19 18:50:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-19 17:02:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-19 17:02:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-19 17:02:38.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-04-19 18:50:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de07963f9490d21aaf17d62003175e6d7491a1ea","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-18 11:22:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cffaed259e936e0ab9e2b6fd43f94ccc73f678c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-18 11:27:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"676ea7e4f52cf0b89eec5523ef2a564edb66fd55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 14:12:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/902a14d454b7489a9b20a47d63f08575 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c24799003fe4d5a944af9ddb9f23131 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0672f3c457214094b572f012b0a9bfcf : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fbbaa49932854bb88a15f0beb3d0dc1f : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1303c7abe2c44054b8b6a8f507afe39c : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f22c174d5fbb46b78c68c38d7e16c12a : SUCCESS in 5m 35s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/5e0d7e698be2456b96cb0637dab94615 : SUCCESS in 5m 29s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2cfb480f27d64497a6e72ea2d138295a : SUCCESS in 2h 44m 12s","accounts_in_message":[],"_revision_number":2},{"id":"46e75a4039c624513aa1b77dbfacaac3d923a4cd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-18 19:09:59.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7996fd1a5e78f29e7307561ff10e93b087dad7e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 21:30:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9d21bc58f5247e3be37c2c570b0082c : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876cda1a0f4446558ebba757137f6c37 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ead2ed4f1ad4114912c21b29ee7f2f2 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abe85992dbd140fe9d18bc3b6e60c74d : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d21df8f244a43a98cbf2e364d651b25 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c865798bab144ed9e020d558c5f626d : SUCCESS in 5m 14s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/d50c96fe4dcf46f99994b1e72390c175 : SUCCESS in 5m 22s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dedad8f20719494198ca789fbae8d578 : SUCCESS in 2h 19m 26s","accounts_in_message":[],"_revision_number":3},{"id":"449067c8b405fc9fe6ef8fe141c87ff239852014","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-19 17:02:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2265c72cac38d1850070e249e6768758711e2e0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 17:03:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5da2505fb625f9730adc7b887f75ee28b36ad51e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 18:50:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"49dbd91265abc20d628cc0d086ad3cc67198256f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 18:50:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5ffb96b88de4f7189700e93e5015678 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5a41047f4c4f4db391beb214fbcc25 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82bc3a9fd82a42bda4279cf50db8f41b : SUCCESS in 2m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40359b308dad4ffc9448bd096b003ae4 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb87b6fc88c4490d888d828c23a158eb : SUCCESS in 4m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/17a2a39053b1472682431ee4fe23a500 : SUCCESS in 5m 50s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/82777323b3ae4cefb9f30deb0529814a : SUCCESS in 1h 46m 27s","accounts_in_message":[],"_revision_number":3},{"id":"083037584810e7160841319dc8e9d53727a622c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-19 18:53:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82343c1438944ed3b796572e4372e78f : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3},{"id":"164902a53e0dbc247a3f33421650bb5edfed1041","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 20:58:59.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 6f0d974ca5587b188335eddc5ceee60ab315bb12","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"359863155811ee82fae1cf9f07d0375d3b61c519","revisions":{"1fb8a3487979e5e0dff68b18bed98c29ab4a83bb":{"kind":"REWORK","_number":1,"created":"2020-04-18 11:22:45.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/80/720880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/80/720880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/1"}}},"commit":{"parents":[{"commit":"e8ac095c897f5c2905981061764fddab7d77cd39","subject":"Merge \"Accept a Row as a result in AddCommand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e8ac095c897f5c2905981061764fddab7d77cd39"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-18 11:19:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-18 11:19:39.000000000","tz":0},"subject":"Tempest job should inherit from Neutron job","message":"Tempest job should inherit from Neutron job\n\nDuring this cycle (Ussuri), OVN driver was integrated as an in-tree\nNeutron driver, deprecating networking-ovn. The tempest job should\nnow inherit from Neutron job definition.\n\nTrivial-Fix\n\nChange-Id: I910ae5d4d32b43d7af15f196f4af6f5e1c4c0734\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1fb8a3487979e5e0dff68b18bed98c29ab4a83bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/1fb8a3487979e5e0dff68b18bed98c29ab4a83bb"}]},"branch":"refs/heads/master"},"556fbb4413693043b455380005e470c9ae3b24ca":{"kind":"REWORK","_number":2,"created":"2020-04-18 11:27:13.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/80/720880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/80/720880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/2"}}},"commit":{"parents":[{"commit":"e8ac095c897f5c2905981061764fddab7d77cd39","subject":"Merge \"Accept a Row as a result in AddCommand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e8ac095c897f5c2905981061764fddab7d77cd39"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-18 11:19:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-18 11:27:05.000000000","tz":0},"subject":"Tempest job should inherit from Neutron job","message":"Tempest job should inherit from Neutron job\n\nDuring this cycle (Ussuri), OVN driver was integrated as an in-tree\nNeutron driver, deprecating networking-ovn. The tempest job should\nnow inherit from Neutron job definition.\n\nTrivial-Fix\n\nChange-Id: I910ae5d4d32b43d7af15f196f4af6f5e1c4c0734\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/556fbb4413693043b455380005e470c9ae3b24ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/556fbb4413693043b455380005e470c9ae3b24ca"}]},"branch":"refs/heads/master"},"359863155811ee82fae1cf9f07d0375d3b61c519":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-18 19:09:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/80/720880/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/80/720880/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/80/720880/3"}}},"commit":{"parents":[{"commit":"e8ac095c897f5c2905981061764fddab7d77cd39","subject":"Merge \"Accept a Row as a result in AddCommand\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e8ac095c897f5c2905981061764fddab7d77cd39"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-18 11:19:39.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-18 19:09:59.000000000","tz":0},"subject":"Tempest job inherits from Neutron tempest OVN job","message":"Tempest job inherits from Neutron tempest OVN job\n\nDuring this cycle (Ussuri), OVN driver was integrated as an in-tree\nNeutron driver, deprecating networking-ovn. The tempest job should\nnow inherit from Neutron job definition.\n\nTrivial-Fix\n\nChange-Id: I910ae5d4d32b43d7af15f196f4af6f5e1c4c0734\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/359863155811ee82fae1cf9f07d0375d3b61c519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/359863155811ee82fae1cf9f07d0375d3b61c519"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
