)]}'
{"id":"openstack%2Fovsdbapp~722346","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0","project":"openstack/ovsdbapp","branch":"stable/train","topic":"bug/1873305-stable/ussuri-stable/train","hashtags":[],"change_id":"Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0","subject":"Allow read-only Commands to bypass txns in execute()","status":"MERGED","created":"2020-04-23 14:55:25.000000000","updated":"2020-05-01 16:14:06.000000000","submitted":"2020-05-01 16:14:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722346-1588349646825-d77804d9","meta_rev_id":"75b40acd0d1c7ec87f552c7bd3c1c24745971686","_number":722346,"virtual_id_number":722346,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-01 16:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":2,"date":"2020-04-23 15:35:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":1,"date":"2020-05-01 14:45:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-01 14:45:54.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-05-01 16:14:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da4146c2db24b1c7cfe422f6ed67f5d0d735e952","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-23 14:55:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"5aee4103de2706fc85dd1e842f07c0ea3c423ce2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-23 14:55:33.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 815d8a392b438490cb9f8132360b76016baa0c22","accounts_in_message":[],"_revision_number":1},{"id":"55b9afbefb777729e43dc0e0585c18620e7b0fd2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-23 15:35:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5afa0a9214b4aa7ae1834402ec122b5a32fc77cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:18:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff31e3fd97224da2bb5bdd16d96d91f1 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0a3162c49cd4bc89a17c3b1a8451382 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e427215f8114a5aba9682b0d1373a78 : SUCCESS in 7m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f306abba0a4646da8b9ca37a82222cba : SUCCESS in 7m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b766d12125c448e78a6d04c9e8705f10 : SUCCESS in 8m 43s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f3eb3800c53c4ff0b663a1d82a9b66c7 : SUCCESS in 8m 49s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/241b2f91d1784743803ec41a74d15c4b : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a1eb56b0a9569d7e113bd601d1560cac95a4a3d6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-01 14:45:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ac6d47f052b0366cf534f6bda1cfa1f4ecb67142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 14:46:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7786570d3cd57c034956e8038a2a8b3931da1d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 16:14:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c00ea5cfc94c0faec640d8c85fb2a29bb0f7322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 16:14:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb11f0ec42c04eceb85327bc454f980d : SUCCESS in 6m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d0de08e2390840c58dadd46397c7ca47 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54feec57d30f4d2984c595baa7d64dc7 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bc0e27a85f14c90b3b02462e7837541 : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe01c829a3c347df845ae30125e6965b : SUCCESS in 5m 36s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8a2873220bc34ee0a57087870aada1d1 : SUCCESS in 8m 28s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/87ad50fa5bad4584bbc2d68bb13b6abd : SUCCESS in 1h 22m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9bdc68b49d04b23009be907656e129dcf753d65","revisions":{"b9bdc68b49d04b23009be907656e129dcf753d65":{"kind":"REWORK","_number":1,"created":"2020-04-23 14:55:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/722346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/46/722346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/46/722346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/46/722346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/46/722346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/46/722346/1"}}},"commit":{"parents":[{"commit":"fd0629f140fe34d24b567ec6d6a10a833f826ce9","subject":"Accept a Row as a result in AddCommand","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/fd0629f140fe34d24b567ec6d6a10a833f826ce9"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-16 15:55:25.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-23 14:55:25.000000000","tz":0},"subject":"Allow read-only Commands to bypass txns in execute()","message":"Allow read-only Commands to bypass txns in execute()\n\nThis patch continues https://review.opendev.org/#/c/633070/.\n\nSince the ovs_idl backend maintains an in-memory copy of the DB,\nread-only commands can be executed w/o creating a transaction.\nSince it is possible for a function that calls execute() to be\nautomatically nested into an existing transaction, execute()\nwasn\u0027t able to access the result since it only happens once the\nparent transaction completed. Handling these read-only commands\nw/o creating a transaction allows the results to be read and\navoids wasting resources by creating needless transactions.\n\nReadOnlyCommands can still be used in with transaction(): blocks\nwith no changes.\n\nChange-Id: Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0\nCloses-Bug: #1873305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b9bdc68b49d04b23009be907656e129dcf753d65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b9bdc68b49d04b23009be907656e129dcf753d65"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
