)]}'
{"id":"openstack%2Fovsdbapp~722347","triplet_id":"openstack%2Fovsdbapp~stable%2Fstein~Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0","project":"openstack/ovsdbapp","branch":"stable/stein","topic":"bug/1873305-stable/ussuri-stable/train-stable/stein","hashtags":[],"change_id":"Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0","subject":"Allow read-only Commands to bypass txns in execute()","status":"MERGED","created":"2020-04-23 14:55:33.000000000","updated":"2020-05-01 16:14:07.000000000","submitted":"2020-05-01 16:14:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722347-1588349647482-6bb723e0","meta_rev_id":"4b84e58fd6e7b524714de143511ce369272dda37","_number":722347,"virtual_id_number":722347,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-01 16:14:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-01 14:46:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-01 14:46:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-01 14:46:30.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-05-01 16:14:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bfacc545af999d625ab1f06b6fd963993165e15d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-23 14:55:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"3217344b3533b7ae7dc4633b99d879424cb6040b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 16:43:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb6094f361e4129b57b03ff177f740a : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93e625f4b7fb493aa18ce84f855b6cdb : FAILURE in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c1cdeacce24a45c2bce872f874a778f7 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/646dc6971cd840ee87a5a666b5534559 : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/610b9bd9f0f64dcb91024cd279a29192 : SUCCESS in 6m 15s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ca3308424f3344b69ad16264e86160c1 : SUCCESS in 5m 41s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/f9d3668f23b646e39908ccc42600b634 : SUCCESS in 1h 40m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1c882e6e9e2115466399eb68f9c9514c04f4c2a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-04-24 09:51:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b68d37da3da059c6701a9c8dd75fbf9ccbd4dc56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 11:07:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5341fbe51854676a438f936c5b75d92 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7a7fa9a8749c4effb83b596da15f2f60 : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d333dc2602d64a2fbe77c344d4b55cf3 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e52d5af0c7dd4cd280383f412230932c : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fda0bd516cd4dc1966375bcb01846d9 : SUCCESS in 8m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8bda10d691964b73b0ec99ae88163667 : SUCCESS in 5m 52s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/589ea166fa5347f495ed8107e4ff1f4d : SUCCESS in 1h 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7eb0fa53784f176dbf294e4852bd0ac13b261362","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-30 04:22:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cf81755057d16ddc3fb4ae1c4da44570e859c518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 05:41:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fe48f5d112f424f84dd77513ce225c3 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0f5c0e24cac5416f8dcd0dfeb0b47ab3 : SUCCESS in 3m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1da960a0063f40c2857264124b6cc993 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66f13cc71f9641b28a7242b0389425fb : SUCCESS in 3m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3af97b9d4f234f9990a4163cf09d8805 : SUCCESS in 4m 38s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ad2a0db660ad45ac90d67486023b750e : SUCCESS in 5m 12s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/6c4aa3d2faf044e4935a4398aaa96851 : SUCCESS in 1h 16m 43s","accounts_in_message":[],"_revision_number":2},{"id":"ce158d5d0d62d8bdce4eef50d6680f259c6e6544","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-01 14:46:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0760276de0f2bc97a4810c506977afc3fa3ad67c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 14:46:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6fdc867dc976e34c95ba9cecc3797c6dd7088a05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 16:14:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/569c0c866b094189bd52777a914b535f : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/02b3fbb263d34062b745cd483f70d266 : SUCCESS in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/daa57e6a864f468e9849fabc25418953 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93604e409d244f0a8723acc235dfa23f : SUCCESS in 4m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9d335a25b234f17a2cd4aa0f4dea753 : SUCCESS in 5m 21s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f24f857900cd4b98b1e41b1b9525f5a3 : SUCCESS in 5m 58s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/2817696a3c274a3c8bec175ee52ec981 : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":2},{"id":"cf2bcd3718b20816eff51ce7e981c2db03e8674f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-01 16:14:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8b8d15b9091717fb2118c6ea664ece508011e0ff","revisions":{"815d8a392b438490cb9f8132360b76016baa0c22":{"kind":"REWORK","_number":1,"created":"2020-04-23 14:55:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/722347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/722347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/722347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/722347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/722347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/722347/1"}}},"commit":{"parents":[{"commit":"9cb971ef46f108e1f1ca9a80ddb92b8c01e4e2ad","subject":"Make Event logging more useful by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9cb971ef46f108e1f1ca9a80ddb92b8c01e4e2ad"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-16 15:55:25.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-23 14:55:33.000000000","tz":0},"subject":"Allow read-only Commands to bypass txns in execute()","message":"Allow read-only Commands to bypass txns in execute()\n\nThis patch continues https://review.opendev.org/#/c/633070/.\n\nSince the ovs_idl backend maintains an in-memory copy of the DB,\nread-only commands can be executed w/o creating a transaction.\nSince it is possible for a function that calls execute() to be\nautomatically nested into an existing transaction, execute()\nwasn\u0027t able to access the result since it only happens once the\nparent transaction completed. Handling these read-only commands\nw/o creating a transaction allows the results to be read and\navoids wasting resources by creating needless transactions.\n\nReadOnlyCommands can still be used in with transaction(): blocks\nwith no changes.\n\nChange-Id: Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0\nCloses-Bug: #1873305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/815d8a392b438490cb9f8132360b76016baa0c22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/815d8a392b438490cb9f8132360b76016baa0c22"}]},"branch":"refs/heads/stable/stein"},"8b8d15b9091717fb2118c6ea664ece508011e0ff":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-30 04:22:52.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/47/722347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/722347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/722347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/722347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/722347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/722347/2"}}},"commit":{"parents":[{"commit":"22a309e76981246da3b329b335c30ada570eda0b","subject":"Call row.verify() when updating map columns with db_set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/22a309e76981246da3b329b335c30ada570eda0b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-16 15:55:25.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-30 04:22:52.000000000","tz":0},"subject":"Allow read-only Commands to bypass txns in execute()","message":"Allow read-only Commands to bypass txns in execute()\n\nThis patch continues https://review.opendev.org/#/c/633070/.\n\nSince the ovs_idl backend maintains an in-memory copy of the DB,\nread-only commands can be executed w/o creating a transaction.\nSince it is possible for a function that calls execute() to be\nautomatically nested into an existing transaction, execute()\nwasn\u0027t able to access the result since it only happens once the\nparent transaction completed. Handling these read-only commands\nw/o creating a transaction allows the results to be read and\navoids wasting resources by creating needless transactions.\n\nReadOnlyCommands can still be used in with transaction(): blocks\nwith no changes.\n\nChange-Id: Ic3f7cd5733b7a28acc74345e7faa182ebd3a69b0\nCloses-Bug: #1873305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8b8d15b9091717fb2118c6ea664ece508011e0ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/8b8d15b9091717fb2118c6ea664ece508011e0ff"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
