)]}'
{"id":"openstack%2Fovsdbapp~723713","triplet_id":"openstack%2Fovsdbapp~stable%2Fstein~Ic898cefb4ced76e9e169b9f5d6caaace600abb0c","project":"openstack/ovsdbapp","branch":"stable/stein","hashtags":[],"change_id":"Ic898cefb4ced76e9e169b9f5d6caaace600abb0c","subject":"Call row.verify() when updating map columns with db_set","status":"MERGED","created":"2020-04-27 21:37:02.000000000","updated":"2020-04-30 03:15:46.000000000","submitted":"2020-04-30 03:15:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723713-1588216546197-3c423189","meta_rev_id":"a49bc6b0723a6b19a808fc2e02ee5aa6183ea6ff","_number":723713,"virtual_id_number":723713,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 03:15:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":2,"date":"2020-04-30 01:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":1,"date":"2020-04-30 01:44:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 03:15:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a469a946573add5daa57decf11a324c06239da50","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-27 21:37:02.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"f40993097c58271e61afa8f5bb3f904c4be34311","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-27 21:37:11.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 9a5547db0629cb3a0ddf696474b6c20b43a4abbb","accounts_in_message":[],"_revision_number":1},{"id":"f981a274b55d8c5b40bccd6ab15a3140fc93b1b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 23:03:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99228ce08fb24d769403ae9929fcb954 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7dd84164d89a407da6a63f61bd19af97 : FAILURE in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aa539891c6504adfae3428bc1c375dcb : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e35637b6d5d452c816479ae876d532d : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1dd474628834e8aaf6e77e1e51ab271 : SUCCESS in 5m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1122a86230be4a70a85181b1aca05f38 : SUCCESS in 5m 53s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/1e3e7325904d4afabf23143355f199eb : SUCCESS in 1h 18m 04s","accounts_in_message":[],"_revision_number":1},{"id":"cefd74cd2532a1fc5bd25ad0fca6253370449dc0","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-28 16:37:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d93cd41ca9d9ba37c13e45ea4473be6ae3dd38fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 18:05:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c20fcbd24964465961afa4f865a44bc : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6edb20e023b4ac88aa82abd39a27c51 : FAILURE in 6m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0a2f4bc024bf47bb9fe2ecb31480b10b : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbe8853b2de343ce81537a16d2f0deb5 : SUCCESS in 8m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8552dc5e8064ab09c2e4a51769661e2 : SUCCESS in 7m 20s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/1157f21092714aefba87369afd8552be : SUCCESS in 9m 14s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/6865ebf4b6ba42f8af50bcf11b43a2cd : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3e5c083c8a16707645d0a80ed54a7cdccc4b04cb","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-29 18:12:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"6c552b61f5245226098b048f418dc2553ed34091","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 19:49:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/273f71bb78094a1096a014be8ea4e5be : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a62b8f09e1b44d9b97cd8b5d14bdf665 : SUCCESS in 4m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dbc32fa5caa547fdb7668004f088222c : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a35e0c92a5b46f39e5b0709565c6ff8 : SUCCESS in 4m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38e45920da144987a9a6fbe5e043a5bf : SUCCESS in 6m 04s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/dc8de5fef5354678804524d2e8b33862 : SUCCESS in 6m 39s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/a4baa221db914e54bfc7779aecde95eb : SUCCESS in 1h 34m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f6416258ba48313a63a25bb483a207135e989fa0","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-04-30 01:44:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f36e044d62552086a04c5259d6f239bf4ba21309","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 01:44:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e7b7fc5ef310d5e597c919043d4fec6cf7b15170","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 03:15:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7efaf4213fd841d2a40b46453b1a81ab : SUCCESS in 3m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/662a42b5f50a47e88571919533144b94 : SUCCESS in 3m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b9309ab342fe49ca93d386217b0ee1b0 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b37c32e26144debafb354f2e8e06553 : SUCCESS in 6m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfd9033e254b425d87758ce21d806baf : SUCCESS in 5m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ebeeeb6bccfd4f35bb4a000d1826a9d0 : SUCCESS in 4m 44s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/12ae565ed5bd43fcb0803fd9d6a28c39 : SUCCESS in 1h 30m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a49bc6b0723a6b19a808fc2e02ee5aa6183ea6ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 03:15:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"22a309e76981246da3b329b335c30ada570eda0b","revisions":{"09119b59a48f73e5f75cb05888f2776d178c16c5":{"kind":"REWORK","_number":1,"created":"2020-04-27 21:37:02.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/13/723713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/13/723713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/13/723713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/13/723713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/13/723713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/13/723713/1"}}},"commit":{"parents":[{"commit":"9cb971ef46f108e1f1ca9a80ddb92b8c01e4e2ad","subject":"Make Event logging more useful by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9cb971ef46f108e1f1ca9a80ddb92b8c01e4e2ad"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-03-16 16:20:05.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-27 21:37:02.000000000","tz":0},"subject":"Call row.verify() when updating map columns with db_set","message":"Call row.verify() when updating map columns with db_set\n\nThe verify() method will ensure that if the row\u0027s column in the db\nhas changed since we set it, that we will be notified and try the\ntransaction again with new values. This is important when we do a\nread-update operation like we do for map columns in db_set. Since\nwe read the values add a few keys, then overwrite the whole value,\nit is important that we don\u0027t do that if we have stale data.\n\nSince ovs 2.6, there has been a setkey() operation that would\nallow modifying a map w/o overwriting the whole value, but there\nis a bug where it doesn\u0027t handle maps that have a refTable value\nlike Qos.queues. So using that before it is fixed [1] and all\npackages have picked it up would cause breakage.\n\n[1] https://patchwork.ozlabs.org/patch/1254735/\n\nCloses-bug: #1867652\nChange-Id: Ic898cefb4ced76e9e169b9f5d6caaace600abb0c\n(cherry picked from commit b92d8742bf3f5e436bcb9160618b343e3c4595fa)\n(cherry picked from commit 5fec1d724461e24206e2c1f66ad6ba8e342d4879)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/09119b59a48f73e5f75cb05888f2776d178c16c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/09119b59a48f73e5f75cb05888f2776d178c16c5"}]},"branch":"refs/heads/stable/stein"},"22a309e76981246da3b329b335c30ada570eda0b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-29 18:12:25.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/13/723713/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/13/723713/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/13/723713/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/13/723713/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/13/723713/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/13/723713/2"}}},"commit":{"parents":[{"commit":"0703dd346877fab5c1ffe7263af04d9fbc4643ac","subject":"Fix spurious test failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/0703dd346877fab5c1ffe7263af04d9fbc4643ac"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-03-16 16:20:05.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-04-29 18:12:25.000000000","tz":0},"subject":"Call row.verify() when updating map columns with db_set","message":"Call row.verify() when updating map columns with db_set\n\nThe verify() method will ensure that if the row\u0027s column in the db\nhas changed since we set it, that we will be notified and try the\ntransaction again with new values. This is important when we do a\nread-update operation like we do for map columns in db_set. Since\nwe read the values add a few keys, then overwrite the whole value,\nit is important that we don\u0027t do that if we have stale data.\n\nSince ovs 2.6, there has been a setkey() operation that would\nallow modifying a map w/o overwriting the whole value, but there\nis a bug where it doesn\u0027t handle maps that have a refTable value\nlike Qos.queues. So using that before it is fixed [1] and all\npackages have picked it up would cause breakage.\n\n[1] https://patchwork.ozlabs.org/patch/1254735/\n\nCloses-bug: #1867652\nChange-Id: Ic898cefb4ced76e9e169b9f5d6caaace600abb0c\n(cherry picked from commit b92d8742bf3f5e436bcb9160618b343e3c4595fa)\n(cherry picked from commit 5fec1d724461e24206e2c1f66ad6ba8e342d4879)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/22a309e76981246da3b329b335c30ada570eda0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/22a309e76981246da3b329b335c30ada570eda0b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
