)]}'
{"id":"openstack%2Fovsdbapp~726235","triplet_id":"openstack%2Fovsdbapp~master~I792a64e7dbadae79e6f12814dc3e0c543998409c","project":"openstack/ovsdbapp","branch":"master","hashtags":[],"change_id":"I792a64e7dbadae79e6f12814dc3e0c543998409c","subject":"Don\u0027t create new API instance in fixtures","status":"MERGED","created":"2020-05-07 20:18:45.000000000","updated":"2020-05-15 11:06:57.000000000","submitted":"2020-05-13 20:16:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726235-1589400993150-2492e8bc","meta_rev_id":"3b954abd341975a5c81cdb29768f0f001afc5843","_number":726235,"virtual_id_number":726235,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 20:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-13 16:09:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"date":"2020-05-13 16:01:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2020-05-13 18:10:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 16:01:31.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-05-13 16:09:01.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-05-13 20:16:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c2f086165c5f691f2168489e2c7f68c3c6ddf81","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-07 20:18:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"786f72a97b9f56575bdbcc8298f60934cfeed371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 22:25:15.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4267d64a931c497493babb1751b13c4d : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0ea615d0bd46aa8fc8e7c2943baf2d : FAILURE in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebf9c5d50c2c4a9998f98324deb59ada : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6fbaaa2f460c4618b3e23776ebf516a6 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/264e8b749e164f7290fb91ec6b374057 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e56f8ec9cd4634a276f6f714a78b32 : SUCCESS in 7m 16s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/53bf2b1cf85f426195ae209664be6f4a : SUCCESS in 6m 15s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/2f39749d09d84b3abfbfa042d7a42044 : SUCCESS in 2h 04m 57s","accounts_in_message":[],"_revision_number":1},{"id":"34d0836ca6e2d731e13adae6af1fafceb76aaef4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-12 16:14:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d83e1521b13565b1fb637a1b4abf81dafba8973e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 18:35:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fab98cfdd28d476495fa44e2d03ed4ab : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/703a591a88754451ad2041874b0cf4ba : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20f68423e46a4f9fb1d39ecb7a98f266 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ad57e6a289f4ff391500959d3f71e9e : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/816ea4cdba354c739f30ee2fcf1fe308 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629c82e449df408fa6da4f7c8249f9a7 : SUCCESS in 6m 55s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/0e245da434454bbd9b4962034c2a4875 : SUCCESS in 10m 55s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/808ce23cda3b48c8847c32687d2802b0 : SUCCESS in 2h 18m 34s","accounts_in_message":[],"_revision_number":2},{"id":"6ef33edc0868bf37e439abb05d73e69b6eee77d4","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-05-13 16:09:01.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"261136e603545524c35f113bc0aaf3c2f7f964fe","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-13 17:27:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8371dead0c298d0391739e3e4133b67ca10040d7","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-13 18:10:53.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"98beef7189ff76842dc1f3878fb16f25b1011d51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:11:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"58a4e0e282ac93d86a936bf363362179e8d526a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 20:16:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a6b7d7c9ab94f94bd0285c6761b8f83 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/279c105bd6f340458c315519d0397eac : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8bbae366b2a450e937c8415629317e7 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/169d8fe603ee4dc49d4afaa297ea08fb : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950b51569914414e969282dd5dd15be3 : SUCCESS in 7m 22s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/555ffb8907904ff6b0181bf3f4a7d7fa : SUCCESS in 8m 34s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a35c452e38e94763910a231e86b38013 : SUCCESS in 1h 57m 02s","accounts_in_message":[],"_revision_number":2},{"id":"04d03c6675ff800c9bc106073d6f7374de360b44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 20:16:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b6d63ff04b21322892a56cdc87196df738de1b31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 20:18:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84db37c4dcff422bb081f072576caf7a : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3b954abd341975a5c81cdb29768f0f001afc5843","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-05-15 11:06:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","revisions":{"2918d586c52b20e3360206ad53cac3c710c4104a":{"kind":"REWORK","_number":1,"created":"2020-05-07 20:18:45.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/35/726235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/35/726235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/35/726235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/35/726235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/35/726235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/35/726235/1"}}},"commit":{"parents":[{"commit":"f8235d80225b7fcbb539679cb3d6fe2c585029ae","subject":"Merge \"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f8235d80225b7fcbb539679cb3d6fe2c585029ae"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 20:12:47.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 20:18:36.000000000","tz":0},"subject":"Don\u0027t create new API instance in fixtures","message":"Don\u0027t create new API instance in fixtures\n\nCreating a new API instance inside each fixture was a little weird.\nNot a big deal as it was relying on the existing class-level\nconnection which was already set. But some users like neutron don\u0027t\nuse class-level connections and explicitly passing the API seems\nless surprising.\n\nChange-Id: I792a64e7dbadae79e6f12814dc3e0c543998409c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2918d586c52b20e3360206ad53cac3c710c4104a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2918d586c52b20e3360206ad53cac3c710c4104a"}]},"branch":"refs/heads/master"},"c5e11c659ede3279b704797bc1cd184e04a8c6ce":{"kind":"REWORK","_number":2,"created":"2020-05-12 16:14:01.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/35/726235/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/35/726235/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/35/726235/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/35/726235/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/35/726235/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/35/726235/2"}}},"commit":{"parents":[{"commit":"f8235d80225b7fcbb539679cb3d6fe2c585029ae","subject":"Merge \"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f8235d80225b7fcbb539679cb3d6fe2c585029ae"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 20:12:47.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-12 16:13:37.000000000","tz":0},"subject":"Don\u0027t create new API instance in fixtures","message":"Don\u0027t create new API instance in fixtures\n\nCreating a new API instance inside each fixture was a little weird.\nNot a big deal as it was relying on the existing class-level\nconnection which was already set. But some users like neutron don\u0027t\nuse class-level connections and explicitly passing the API seems\nless surprising.\n\nChange-Id: I792a64e7dbadae79e6f12814dc3e0c543998409c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
