)]}'
{"id":"openstack%2Fovsdbapp~727766","triplet_id":"openstack%2Fovsdbapp~stable%2Fqueens~I7818c5b60eabca08963c623dd131784791b7dc37","project":"openstack/ovsdbapp","branch":"stable/queens","topic":"bug/1868686","hashtags":[],"change_id":"I7818c5b60eabca08963c623dd131784791b7dc37","subject":"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"","status":"MERGED","created":"2020-05-13 13:40:08.000000000","updated":"2020-05-13 17:02:04.000000000","submitted":"2020-05-13 17:02:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727766-1589389324509-eb83fbf0","meta_rev_id":"37d43e4b41223899a0e1830e474884211b535e19","_number":727766,"virtual_id_number":727766,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 17:02:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":2,"date":"2020-05-13 15:23:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":1,"date":"2020-05-13 15:23:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 15:23:41.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-05-13 17:02:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5998fe7f21a363c018a794570f68fe4fa80e8b43","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-13 13:40:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d9f702c7141beb89db2245be9bd68bfcd6d452e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 14:54:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d026b57bba4749879b554458a38699df : SUCCESS in 4m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97949dee2d614fe89d3027099f92ce51 : SUCCESS in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a5d957de9f3f40cd87641f8645ad0a73 : SUCCESS in 5m 19s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9c752203d8924a278d9a58e150156999 : SUCCESS in 6m 36s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/65ece6b250554e01ac0966dbfc051d09 : SUCCESS in 1h 06m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6859745905ac3c61e6ee38eaf7b31426defaecf3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-13 15:23:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"41e0d38d91525c92338600c360dd2f6ee1f8b3e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 15:23:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03c6a9a40b5357b2ae6e361304d09bb50cf0d04f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 17:02:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"58cc1216723f6b4ddf8cd9a091e5255422e830c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 17:02:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34799a1106de4410a20db141387d11c1 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e4f2b3c892a48e2aad46e02846edac2 : SUCCESS in 6m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2786a1b12c834082b15ac72b662794fc : SUCCESS in 4m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a73c2135877b4c79804cec41fad04c89 : SUCCESS in 5m 32s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/f86aa7a2d0cd491aa90118763fb7a5c5 : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1516c706964f848fe631caaf8e473d9afb5406c","revisions":{"d1516c706964f848fe631caaf8e473d9afb5406c":{"kind":"REWORK","_number":1,"created":"2020-05-13 13:40:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/727766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/66/727766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/66/727766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/66/727766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/66/727766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/66/727766/1"}}},"commit":{"parents":[{"commit":"29e2f7bbab48b8902179cd323d74654965221449","subject":"Merge \"Call row.verify() when updating map columns with db_set\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/29e2f7bbab48b8902179cd323d74654965221449"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-04-17 17:23:47.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 13:40:01.000000000","tz":0},"subject":"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"","message":"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"\n\nAdded \"inactivity_probe\" parameter (milliseconds) to the set manager\ncommand. The command timeout and the manager inactivity probe are\nnow two different parameters.\n\nThis patch also solves a problem related with the kwargs passed to\nthe command executor. Both related parameters, \"timeout\" and\n\"inactivity_probe\" are removed first from this dictionary.\n\nThis function is, at the present time, only called once from Neutron.\n\nChange-Id: I7818c5b60eabca08963c623dd131784791b7dc37\nRelated-Bug: #1868686\n(cherry picked from commit 3ff1e33f2b3fb9e50acfdf60f82a962fc2c94836)\n(cherry picked from commit bdf697a1a02318b1017f7a3bc16e0c187230fada)\n(cherry picked from commit 9bb0eee28608d554076a1d9273c607afe0a5e0c3)\n(cherry picked from commit 1cb33fa2d04d565683911fcbd4731e585bd5d5b3)\n(cherry picked from commit 1bc0f9f1715a9051492948314aa69259e822768a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d1516c706964f848fe631caaf8e473d9afb5406c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/d1516c706964f848fe631caaf8e473d9afb5406c"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
