)]}'
{"id":"openstack%2Fovsdbapp~727847","triplet_id":"openstack%2Fovsdbapp~master~I0eb7151eabaf40232f79797b2775c9bb7891013f","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1877408","hashtags":[],"change_id":"I0eb7151eabaf40232f79797b2775c9bb7891013f","subject":"Add QoSDelExtIdCommand","status":"MERGED","created":"2020-05-13 17:33:29.000000000","updated":"2020-06-23 13:02:25.000000000","submitted":"2020-06-23 13:00:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":64,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727847-1592917217948-0147b0e7","meta_rev_id":"e49fd9596d6e7371c36bb3680ddb4f0c3600220c","_number":727847,"virtual_id_number":727847,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-23 13:00:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-22 19:09:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":2,"date":"2020-06-23 11:02:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":1,"date":"2020-06-23 11:02:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 10:36:36.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2020-06-16 09:49:29.000000000","updated_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"reviewer":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"state":"REVIEWER"},{"updated":"2020-06-22 19:09:33.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2020-06-23 11:02:12.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-06-23 13:00:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac2b82a7c02a065716a1736c528c869539db438c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-13 17:33:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbf1ab7bf28182391349efb418851778648f0196","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-13 17:34:09.000000000","message":"Patch Set 1: Workflow-1\n\nWIP","accounts_in_message":[],"_revision_number":1},{"id":"2829cb20fc9daee722e777786beb4c0e039d6eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 20:12:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86629f81b9ac4370a893de99fcfe66dc : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b410923e52d4760bc02c6fe839a093f : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97b932655f524f158aebc0877a92da24 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7752ece26514bfb9254ccb5fc091b18 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d02c6a9d69ce439d85174c08151f11fa : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa931cdc77e4c32a7c768aaed117048 : SUCCESS in 8m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f13500636ca441fbb4d63549faa833d9 : SUCCESS in 8m 36s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/26f8d2c3450f41299ebe110b36814ad9 : SUCCESS in 2h 32m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a7d8ffdc594ac8bc2fd30e7e2ecca9609b2a7695","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-14 10:33:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c08df8ea744da92b3dd10c1fd3018cceff39474","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 12:12:52.000000000","message":"Patch Set 2: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4492f2224df845098e91b9a6954e5558 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869da55f991a43c2b957bea6b72350a2 : FAILURE in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ab4deaac11c4ba6b16c94d8078e621c : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/839125a652334416b3b9c8b1cf9c70f8 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8257f0ddb6a244bf8b5497bf7fe90aa5 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3873d01f7bc541c099d69446060aca8d : SUCCESS in 6m 32s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/ee31a77f927c438c89efc0e0f309b83a : SUCCESS in 7m 27s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/24374b37bbdc4d2496d21b121a64a466 : SUCCESS in 1h 37m 46s","accounts_in_message":[],"_revision_number":2},{"id":"7dd0abd453a6e97931718061fd2952a9f5f8a95b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-14 12:57:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"49455e8885eeca0fd11f0111fe682286294fea74","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-14 14:22:07.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"87242d899ac1e294ca405981cd23ff658b93ad16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 16:27:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5eeefe6a6b6f48b4b897e9b34d734333 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75af71e2770e4a709f90dc9354fa8834 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50a1f8be14054b36a4f7131f953d898c : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2dbdb1df6e244e00a507ea5f9fa60e8b : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/710f9f887a3b466f88f3ec27abd5eb33 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f821c39c81f44dc8be9ac10c12218437 : SUCCESS in 5m 02s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/817c0f7513e945f886045df14a1a0fea : SUCCESS in 7m 20s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0ada085363164963a69e7c260853bee0 : SUCCESS in 1h 58m 45s","accounts_in_message":[],"_revision_number":4},{"id":"3a8e269a04f2a155e83266896b20591e0e2b2316","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-15 20:15:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nThanks! Mostly looks good, just a couple of small questions.\n\nOne additional question: Is this how you would expect a non-neutron user of ovsdbapp to handle FIPs and QoS? Is this generally applicable to everyone, or is it neutron-specific? I see that you allow passing in the ext_id_key, so you\u0027ve clearly thought about it. :D My gut feeling would be \"well, external-ids are there to be used for this kind of thing, so it seems like what anyone would do\" but I haven\u0027t looked hard enough to really know. If it is neutron-specific, we\u0027d put it in the neutron tree. Otherwise, I\u0027m quite happy for it to live here.","accounts_in_message":[],"_revision_number":4},{"id":"0ef879959c1869ae2ff6572e0574529aba7657ec","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-05-18 07:47:07.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d3d69cf198c85c76d44212aba50eaf73e060bcf9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-18 09:15:03.000000000","message":"Patch Set 4:\n\n\u003e (4 comments)\n \u003e \n \u003e Thanks! Mostly looks good, just a couple of small questions.\n \u003e \n \u003e One additional question: Is this how you would expect a non-neutron\n \u003e user of ovsdbapp to handle FIPs and QoS? Is this generally\n \u003e applicable to everyone, or is it neutron-specific? I see that you\n \u003e allow passing in the ext_id_key, so you\u0027ve clearly thought about\n \u003e it. :D My gut feeling would be \"well, external-ids are there to be\n \u003e used for this kind of thing, so it seems like what anyone would do\"\n \u003e but I haven\u0027t looked hard enough to really know. If it is\n \u003e neutron-specific, we\u0027d put it in the neutron tree. Otherwise, I\u0027m\n \u003e quite happy for it to live here.\n\nYou are right, we need to make this as generic as possible. That means we should not:\n- implement a specific logic only used in Neutron (although I was also passing the key name, instead of hardcoding it)\n- call those new classes with the Neutron logic (floating IP in this case)\n\nMy idea:\n\n1) I\u0027ve seen that QoSDelCommand is used only in ovsdbapp, that means we can \"freely\" change the signature and improve the functionality. My idea is to, instead of creating a new command, improve QoSDelCommand.\n\n2) About the API, because \"qos_del\" is already defined and we can\u0027t know where is used, we can implement a new method, more generic. Something like \"qos_del_ext_id\"\n\nWhat do you think?\n\n[1]http://codesearch.openstack.org/?q\u003dQoSDelCommand\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d","accounts_in_message":[],"_revision_number":4},{"id":"90d585893d2b4bb8a2643da95d42fa6ad3af0d3c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-05-18 09:17:57.000000000","message":"Patch Set 4:\n\n(3 comments)\n\n\u003e \u003e (4 comments)\n \u003e \u003e\n \u003e \u003e Thanks! Mostly looks good, just a couple of small questions.\n \u003e \u003e\n \u003e \u003e One additional question: Is this how you would expect a\n \u003e non-neutron\n \u003e \u003e user of ovsdbapp to handle FIPs and QoS? Is this generally\n \u003e \u003e applicable to everyone, or is it neutron-specific? I see that you\n \u003e \u003e allow passing in the ext_id_key, so you\u0027ve clearly thought about\n \u003e \u003e it. :D My gut feeling would be \"well, external-ids are there to\n \u003e be\n \u003e \u003e used for this kind of thing, so it seems like what anyone would\n \u003e do\"\n \u003e \u003e but I haven\u0027t looked hard enough to really know. If it is\n \u003e \u003e neutron-specific, we\u0027d put it in the neutron tree. Otherwise, I\u0027m\n \u003e \u003e quite happy for it to live here.\n \u003e \n \u003e You are right, we need to make this as generic as possible. That\n \u003e means we should not:\n \u003e - implement a specific logic only used in Neutron (although I was\n \u003e also passing the key name, instead of hardcoding it)\n \u003e - call those new classes with the Neutron logic (floating IP in\n \u003e this case)\n \u003e \n \u003e My idea:\n \u003e \n \u003e 1) I\u0027ve seen that QoSDelCommand is used only in ovsdbapp, that\n \u003e means we can \"freely\" change the signature and improve the\n \u003e functionality. My idea is to, instead of creating a new command,\n \u003e improve QoSDelCommand.\n \u003e \n \u003e 2) About the API, because \"qos_del\" is already defined and we can\u0027t\n \u003e know where is used, we can implement a new method, more generic.\n \u003e Something like \"qos_del_ext_id\"\n \u003e \n \u003e What do you think?\n \u003e \n \u003e [1]http://codesearch.openstack.org/?q\u003dQoSDelCommand\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nHehe, I thought about it too even without reading the comment :) I thought about just renaming the class to something more generic like QoSDelExtId and rename \"fip\" to \"value\" or something generic.","accounts_in_message":[],"_revision_number":4},{"id":"2ce263f9de12ecbe47e74849ec7c78bd4f766254","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-18 13:25:02.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7dfca1a0cad9c313ee5c9176c2de51ad4db6b77f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-18 14:02:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2663afe5c3d96a45e2afdf0d7a8fdff607bc2104","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-18 14:07:55.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"ba4dee24c8e499ba1044abf80c632587923f7db4","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-18 15:29:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"49a3d1e16e1516d30b07f0d67870bc05de9f14f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 16:37:41.000000000","message":"Patch Set 6: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/021249c67fc7418ebbf19cf2c4569f2f : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/277400a2fea442d5b0dcb91fd6e0c754 : FAILURE in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bd75c8607e24da7a41bd29cb06672a0 : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d906546f76ae4fb19c00060c44c46817 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66e47c67725c45f7992d58e48f55e1bb : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/178ee48700e847c8a487adc6b37f38ef : SUCCESS in 6m 39s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e7bcd92ab3b54a1b960a3fd9d680f284 : SUCCESS in 5m 50s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4f4c0f8ae6034bd99144f63d5de90bc8 : SUCCESS in 2h 26m 45s\n\nWarning:\n  Comments left for invalid file ovsdbapp/tests/functional/schema/open_vswitch/test_common_db.py","accounts_in_message":[],"_revision_number":6},{"id":"36f0aa1a91c02fec79337dc58b0eacb74a183d6e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 22:04:26.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e6ce69afe589d4c23c2fe0a6168fce7e6a02ca28","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-20 10:36:15.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b6b15b0b49fe7bb630b55685e9780db84d74c8c5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-20 13:05:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3a79ebc07ac730805073c4e0fe059d3e06d2bb32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 15:30:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db9c8d07d20c488c8c135354394a4447 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a298b5315b4eb0a7be57c12dd0d24c : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c0b29a9df444f1791ce422c1c4ba524 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d61e9fbf61bc4d47a316901376609531 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7379c882c0f4f93b66e3df340c3e952 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db42519f04da4837b981ed240a2d465e : SUCCESS in 5m 33s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/40ac765bea974216b8c1cc4894534db8 : FAILURE in 9m 50s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/95f1f5fbd81e4a96b601aa4bec76040a : SUCCESS in 2h 17m 31s","accounts_in_message":[],"_revision_number":7},{"id":"e1b0a9b91d9e20b99fe3a34f3bb7f892d932eb3b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-20 17:42:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5d0270327b05b211e420adf8edf1839d515614c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 19:51:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9cbf12604ba4e26b64f0d31d3510da9 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/639eb063ccfd429899a40cfa8d21c8e0 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/164025fdc97249ea8c8e5f76a985f3af : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5783800b23a449cba836033d0ecff519 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71602fa8510c403e93f07c9d63522ce4 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2937eec1fe0547b182d62dbcfc85f545 : SUCCESS in 7m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/784e9801b58e4a00a9ba3c6004942962 : SUCCESS in 6m 08s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/175208619f6e48c0a091a3b34f66c542 : SUCCESS in 2h 02m 56s","accounts_in_message":[],"_revision_number":8},{"id":"150320d5b5c969f4a1240566bee49f7b7f92ceb9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-26 08:46:46.000000000","message":"Patch Set 8:\n\nping fellow reviewers","accounts_in_message":[],"_revision_number":8},{"id":"d4464c51bf0f4c927cad3acbef2b38f40b01aba3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-08 10:36:19.000000000","message":"Patch Set 8:\n\nping folks","accounts_in_message":[],"_revision_number":8},{"id":"987060d6ce5466a87dd9faac526c7bf3d792a7f2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-06-09 19:19:54.000000000","message":"Patch Set 8: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4af414fb1c061f5705592d01041952b0a12ba6d1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-10 11:25:44.000000000","message":"Patch Set 8:\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3d731497ff34897a6e85de6ec359631014a653af","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-10 11:36:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1e16024b288300713a69f3964f6a5a5c6c24d92f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 13:26:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/782407d9f9df4b4da6cd40c2ab51ddcb : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f5a7d9ab3c64c2489c1aa963a81b952 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3992df7661ad40b496138f59b7fa6873 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b4bd50f9eac47f2bf587bb63d9972f5 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64dd5e7f8ff74b7fad74a5b233643af2 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64d39c21aa94cb58667c22be6508e01 : SUCCESS in 5m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/48c260b7894c4ada9d80eb9336ac6802 : SUCCESS in 6m 03s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4df60c296a28494e8636c9516a320384 : SUCCESS in 1h 48m 22s","accounts_in_message":[],"_revision_number":9},{"id":"44996e4c91f6fece88447abe9ee3b148a6b3a03b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-06-10 13:47:39.000000000","message":"Patch Set 9: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":9},{"id":"179d5d76017c3779e85dd15ccf0ce43eda38535e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-16 09:36:56.000000000","message":"Patch Set 9:\n\nHi folks, do you have some time for this patch? Thanks!","accounts_in_message":[],"_revision_number":9},{"id":"91878fda2ebab9122e421b928ba439cc2bde9bde","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-06-16 09:49:29.000000000","message":"Patch Set 9: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":9},{"id":"41095d4d2b364a0e7cd115954f453e75db6663e2","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-06-16 10:15:22.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"16adb018f4b927dd1d94718e884f268465c7efc7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-16 18:02:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5217827b04bffb7f9bb734c838a2395e7f599c89","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-16 18:56:57.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a89e5a31e5d65f97a87e51c55b9e768fd5d364bb","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-06-17 12:49:35.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)\n\nThanks for explanation, I learned more about dict views :)","accounts_in_message":[],"_revision_number":9},{"id":"657996fe4a3bd9bb7df2fde2329b1277b3e2acee","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-06-17 12:51:04.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7800e17f409bd80eb762ef3ace472d9d086247c7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-18 09:32:35.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a4cd877e4c3ea851915e60f452c74c3b8e7cc28a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-18 09:32:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"370235c347d17dc1564d651f7f4bebaf7156a482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 09:57:38.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61547f62c4474d74a775a25ee81e61e9 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11c1d7552d564b3f84bea6286b505b21 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2eb1361b50ce40ad9016f1a89195239d : SUCCESS in 5m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b14b4c2a813346758512c2b74e1464a5 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70a6c3034b55410ea0ca289727777c09 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d47c4388e1471391aa324fb0ad082d : SUCCESS in 6m 06s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/8ca4bfdb16e84408ac5982cfa355cb51 : SUCCESS in 6m 56s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3e1d56e4af834aea9aa7d9a9f6236be9 : FAILURE in 18m 56s","accounts_in_message":[],"_revision_number":10},{"id":"442f37825348791f5cb6c39f192f114eba08c29e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-06-18 09:59:30.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)\n\nThanks for implementing the new tests!","accounts_in_message":[],"_revision_number":10},{"id":"b4f9237f48029701b6db7824e5596dc7a88bec6b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-18 14:59:07.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"bf794e37841aeea5840e4ceb02fe0e7215e670e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:36:10.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d396a9ebe324c5394b6d5821043e70d : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25955fed6efe4f8ca3240e4c01f569f3 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/495053e229154e5badae237d1edf02dd : SUCCESS in 7m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7216557216246c3aa04d96249c0fa01 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c362e43f06724629bf8aa87764764f33 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef1ccd01be124e2d99f4c4dde6434abe : SUCCESS in 7m 58s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/2dc6f30932954c54999b9fcb0dcda325 : SUCCESS in 9m 07s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b8ad74cebe9a4bf78eb6814216380ce2 : FAILURE in 26m 17s","accounts_in_message":[],"_revision_number":10},{"id":"4ed2b25b36cca4ff9340b0df15ff238998457e26","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-18 23:39:41.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"c58b36d2a6f85b746ce5f8a4d841d8452e1fa7d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 00:08:46.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a3a75d046764ebd89b53a56711be0db : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa885729ee094f88a873387ca0be12ed : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6af9555709634128b288e9553c087b07 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c25c6a5ef35414ca3f75e82fa2cb1b0 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f47594fe7ce04e019fb5ab0bc7fafa27 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc28503922d48d79dfa14cc4155c6de : SUCCESS in 5m 37s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/fcb6c5aa52a54ceca9174ed687a45a58 : FAILURE in 6m 23s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/73bf631bfb1f4b6aacd17e1556b6f9db : FAILURE in 18m 10s","accounts_in_message":[],"_revision_number":10},{"id":"f00c15171c712a3eeecbb0ee57063d33a389a886","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-19 07:57:33.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"c58ad73ad064a5a51dbd74547b68e2f19e488a08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 08:24:03.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5a475f97acc4dec8ea51a502a31e6c2 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d1d7ae397d7429f9e5f31007f3d1d09 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dfb4f9440074611b66c013c6481fe5c : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3701ff1164042d085f34f977b7e5efe : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0251d6e3e4043fd8c89c614e36de55b : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665f85df0cf044189c288cf1f62efffd : SUCCESS in 5m 40s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/6fd481e23aa3447182b1e18bf7a8d136 : SUCCESS in 7m 46s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b35afdd4c3ea43b59236eec4e4d971ed : FAILURE in 23m 54s","accounts_in_message":[],"_revision_number":10},{"id":"f7bf72509dc19ecff20dc51398d6fec771b692f6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-19 22:38:51.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"8f496b59c3f7780689043504c57a0e4b8e62685d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 00:47:45.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0ecbb5549fa47318c02541291324537 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfcc2ac7b58c45b8bc4e11dff88cfc0d : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47fdccb07aaf43b987d6970df2b0739e : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cefa3530290f419ebfa4b700f8373b83 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3587e998f4ea4a34974032d217635606 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499ff5787e5e4f5cbafacc63d7c6e0ab : SUCCESS in 10m 25s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/f38c257fc6cb455eb1cf71ebc1af670a : SUCCESS in 8m 15s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/99742d12dce742019284e157f1b7b448 : SUCCESS in 2h 06m 17s","accounts_in_message":[],"_revision_number":10},{"id":"4cfeab7eaa9ccee0e7af188971f07a32eb9fc109","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-06-22 12:16:30.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"9f0088e7b0d58e90176c7bb3dbc06b85b55216a5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-22 16:32:39.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"c70f9da0198c00aa615e3922a2a0a43ee7c6d1a9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-22 16:45:25.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"da2056da2fbaa6dab351aa9a2ddaa178b5c890da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 18:25:17.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b51c0a7ab32485aa613f435d792127b : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c5a65069a54dabbec25bb93eafc141 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bad2d82da39540039211099cf780ff7e : SUCCESS in 7m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c7246b7c646241ca8c7fed524f838c31 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09d3f980194642168b21ea2df8b6ffa7 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c06e5e2a6dd4447fb0c5b6a09a86db2d : SUCCESS in 5m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/4b9d0e2508724323ac9b27ad7e379eb4 : SUCCESS in 6m 34s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/537e97f1c0ef47499df0075506427b62 : SUCCESS in 1h 37m 11s","accounts_in_message":[],"_revision_number":11},{"id":"45ea7501c92e89db51556d561790443e92d24c13","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-06-22 19:09:33.000000000","message":"Patch Set 11: Code-Review+2\n\nnice work, thanks!","accounts_in_message":[],"_revision_number":11},{"id":"f19a81b0364b839990e60f6df302912da1279e43","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-06-23 11:02:12.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"cbfa4f9de7022ba0b17cb1c233657c734c332a63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 11:02:23.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"c0f8ad7d48d0a4fdc1387d3cd6c3fc387b9f8276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 13:00:17.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0b828d818d24d4c8b81a3c3e7308632 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a438ce56148b4c9387af6fe1b52d2908 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f947112d6b464eb5b2dff6afe726b761 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c10834adfca4d45922f541657b533a4 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a34f0710474444af4387c35e304f26 : SUCCESS in 6m 49s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a75c8ee3bde34d7bafe9348c883ffd3b : SUCCESS in 7m 15s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f59691b863e64faaad15bc730f44dbdf : SUCCESS in 1h 56m 20s","accounts_in_message":[],"_revision_number":11},{"id":"c5e3132ee27ed3349643559fb5f6398f1b73119a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 13:00:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"32498bd38f0eb404ad448aaa600c41246aacb4e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 13:02:25.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08cca958074b4c14b4dab8d36e9d6772 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"08605e94a411d3dabcff63f40b71d6e34bb3ed72","revisions":{"a6daeabfaedb04338afbcb4597851df6fd084327":{"kind":"REWORK","_number":1,"created":"2020-05-13 17:33:29.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/1"}}},"commit":{"parents":[{"commit":"f8235d80225b7fcbb539679cb3d6fe2c585029ae","subject":"Merge \"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f8235d80225b7fcbb539679cb3d6fe2c585029ae"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"subject":"Add QoSFipDelCommand","message":"Add QoSFipDelCommand\n\nThis command removes all QoS rules related to a floating IP. Those\nQoS registers are retrieved from the DB filtering by\n  \"ext_id_key\" \u003d\u003d \"floatingip_id\"\n\nIn Neutron, \"ext_id_key\" is\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a6daeabfaedb04338afbcb4597851df6fd084327"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/a6daeabfaedb04338afbcb4597851df6fd084327"}]},"branch":"refs/heads/master"},"9b8655a43d7d0af69d55834df7e0da20720962de":{"kind":"REWORK","_number":2,"created":"2020-05-14 10:33:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/2"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-14 10:33:28.000000000","tz":0},"subject":"Add QoSFipDelCommand","message":"Add QoSFipDelCommand\n\nThis command removes all QoS rules related to a floating IP. Those\nQoS registers are retrieved from the DB filtering by\n  \"ext_id_key\" \u003d\u003d \"floatingip_id\"\n\nIn Neutron, \"ext_id_key\" is\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9b8655a43d7d0af69d55834df7e0da20720962de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9b8655a43d7d0af69d55834df7e0da20720962de"}]},"branch":"refs/heads/master"},"7ac532453902dc79018a24cd992ae27cbecc5f47":{"kind":"REWORK","_number":3,"created":"2020-05-14 12:57:12.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/3"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-14 12:57:01.000000000","tz":0},"subject":"Add QoSFipDelCommand","message":"Add QoSFipDelCommand\n\nThis command removes all QoS rules related to a floating IP. Those\nQoS registers are retrieved from the DB filtering by\n  \"ext_id_key\" \u003d\u003d \"floatingip_id\"\n\nIn Neutron, \"ext_id_key\" is\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7ac532453902dc79018a24cd992ae27cbecc5f47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7ac532453902dc79018a24cd992ae27cbecc5f47"}]},"branch":"refs/heads/master"},"2ad39b2974f3a70316736e58c95a3bc0fc46cc3e":{"kind":"REWORK","_number":4,"created":"2020-05-14 14:22:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/4"}}},"commit":{"parents":[{"commit":"c5e11c659ede3279b704797bc1cd184e04a8c6ce","subject":"Don\u0027t create new API instance in fixtures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/c5e11c659ede3279b704797bc1cd184e04a8c6ce"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-14 14:22:05.000000000","tz":0},"subject":"Add QoSFipDelCommand","message":"Add QoSFipDelCommand\n\nThis command removes all QoS rules related to a floating IP. Those\nQoS registers are retrieved from the DB filtering by\n  \"ext_id_key\" \u003d\u003d \"floatingip_id\"\n\nIn Neutron, \"ext_id_key\" is\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ad39b2974f3a70316736e58c95a3bc0fc46cc3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/2ad39b2974f3a70316736e58c95a3bc0fc46cc3e"}]},"branch":"refs/heads/master"},"f31f3eb89be0dbd70b2ca6343f57f029e956c5fc":{"kind":"REWORK","_number":5,"created":"2020-05-18 14:02:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/5"}}},"commit":{"parents":[{"commit":"04f29abfe026f5c2d56b289643bb2cac89898904","subject":"Change \"fixture\" module names in testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/04f29abfe026f5c2d56b289643bb2cac89898904"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-18 14:01:59.000000000","tz":0},"subject":"Add QoSFipDelCommand","message":"Add QoSFipDelCommand\n\nThis command removes all QoS rules related to a floating IP. Those\nQoS registers are retrieved from the DB filtering by\n  \"ext_id_key\" \u003d\u003d \"floatingip_id\"\n\nIn Neutron, \"ext_id_key\" is\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f31f3eb89be0dbd70b2ca6343f57f029e956c5fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/f31f3eb89be0dbd70b2ca6343f57f029e956c5fc"}]},"branch":"refs/heads/master"},"e9f7f3ee240d256818b4fac20b8cd80f62250c52":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-05-18 14:07:55.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/6"}}},"commit":{"parents":[{"commit":"04f29abfe026f5c2d56b289643bb2cac89898904","subject":"Change \"fixture\" module names in testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/04f29abfe026f5c2d56b289643bb2cac89898904"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-18 14:02:42.000000000","tz":0},"subject":"Add QoSDelExtIdCommand","message":"Add QoSDelExtIdCommand\n\nThis command removes all QoS rules matching a set of keys and\nexpected values passed in the command instantiation. No exception\nis thrown if no QoS is found with the required external IDs.\n\nFor example, this will be used in Neutron to delete all QoS\nregisters related to a floating IP, searching for\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY key and\nthe floating IP register UUID.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e9f7f3ee240d256818b4fac20b8cd80f62250c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e9f7f3ee240d256818b4fac20b8cd80f62250c52"}]},"branch":"refs/heads/master"},"e792b859fcd569ee71d3d590eb3ce438ddbd8941":{"kind":"REWORK","_number":7,"created":"2020-05-20 13:05:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/7"}}},"commit":{"parents":[{"commit":"7736bac7dc093abbcb6990cf84d46c2fb6beeb9d","subject":"Merge \"Add indexed column support to ovsdbapp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7736bac7dc093abbcb6990cf84d46c2fb6beeb9d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-20 13:05:38.000000000","tz":0},"subject":"Add QoSDelExtIdCommand","message":"Add QoSDelExtIdCommand\n\nThis command removes all QoS rules matching a set of keys and\nexpected values passed in the command instantiation. No exception\nis thrown if no QoS is found with the required external IDs.\n\nFor example, this will be used in Neutron to delete all QoS\nregisters related to a floating IP, searching for\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY key and\nthe floating IP register UUID.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e792b859fcd569ee71d3d590eb3ce438ddbd8941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/e792b859fcd569ee71d3d590eb3ce438ddbd8941"}]},"branch":"refs/heads/master"},"b56e65b8aa82107d2b4482db19a6b733529cd21f":{"kind":"REWORK","_number":8,"created":"2020-05-20 17:42:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/8"}}},"commit":{"parents":[{"commit":"7736bac7dc093abbcb6990cf84d46c2fb6beeb9d","subject":"Merge \"Add indexed column support to ovsdbapp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7736bac7dc093abbcb6990cf84d46c2fb6beeb9d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-20 17:41:49.000000000","tz":0},"subject":"Add QoSDelExtIdCommand","message":"Add QoSDelExtIdCommand\n\nThis command removes all QoS rules matching a set of keys and\nexpected values passed in the command instantiation. No exception\nis thrown if no QoS is found with the required external IDs.\n\nFor example, this will be used in Neutron to delete all QoS\nregisters related to a floating IP, searching for\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY key and\nthe floating IP register UUID.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b56e65b8aa82107d2b4482db19a6b733529cd21f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b56e65b8aa82107d2b4482db19a6b733529cd21f"}]},"branch":"refs/heads/master"},"6d3274a07977e680943f440bcce9d9ab96fe7b6a":{"kind":"REWORK","_number":9,"created":"2020-06-10 11:36:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/9"}}},"commit":{"parents":[{"commit":"7736bac7dc093abbcb6990cf84d46c2fb6beeb9d","subject":"Merge \"Add indexed column support to ovsdbapp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7736bac7dc093abbcb6990cf84d46c2fb6beeb9d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-10 11:36:38.000000000","tz":0},"subject":"Add QoSDelExtIdCommand","message":"Add QoSDelExtIdCommand\n\nThis command removes all QoS rules matching a set of keys and\nexpected values passed in the command instantiation. No exception\nis thrown if no QoS is found with the required external IDs.\n\nFor example, this will be used in Neutron to delete all QoS\nregisters related to a floating IP, searching for\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY key and\nthe floating IP register UUID.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6d3274a07977e680943f440bcce9d9ab96fe7b6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/6d3274a07977e680943f440bcce9d9ab96fe7b6a"}]},"branch":"refs/heads/master"},"12aa519ce475c6bfa6ba48e2a4b4ee7c3af751c7":{"kind":"REWORK","_number":10,"created":"2020-06-18 09:32:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/10"}}},"commit":{"parents":[{"commit":"7736bac7dc093abbcb6990cf84d46c2fb6beeb9d","subject":"Merge \"Add indexed column support to ovsdbapp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7736bac7dc093abbcb6990cf84d46c2fb6beeb9d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-18 09:32:22.000000000","tz":0},"subject":"Add QoSDelExtIdCommand","message":"Add QoSDelExtIdCommand\n\nThis command removes all QoS rules matching a set of keys and\nexpected values passed in the command instantiation. No exception\nis thrown if no QoS is found with the required external IDs.\n\nFor example, this will be used in Neutron to delete all QoS\nregisters related to a floating IP, searching for\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY key and\nthe floating IP register UUID.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/12aa519ce475c6bfa6ba48e2a4b4ee7c3af751c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/12aa519ce475c6bfa6ba48e2a4b4ee7c3af751c7"}]},"branch":"refs/heads/master"},"08605e94a411d3dabcff63f40b71d6e34bb3ed72":{"kind":"REWORK","_number":11,"created":"2020-06-22 16:45:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/47/727847/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/47/727847/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/47/727847/11"}}},"commit":{"parents":[{"commit":"7736bac7dc093abbcb6990cf84d46c2fb6beeb9d","subject":"Merge \"Add indexed column support to ovsdbapp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7736bac7dc093abbcb6990cf84d46c2fb6beeb9d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-13 17:30:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-06-22 16:44:40.000000000","tz":0},"subject":"Add QoSDelExtIdCommand","message":"Add QoSDelExtIdCommand\n\nThis command removes all QoS rules matching a set of keys and\nexpected values passed in the command instantiation. No exception\nis thrown if no QoS is found with the required external IDs.\n\nFor example, this will be used in Neutron to delete all QoS\nregisters related to a floating IP, searching for\nneutron.common.ovn.constants.OVN_FIP_EXT_ID_KEY key and\nthe floating IP register UUID.\n\nChange-Id: I0eb7151eabaf40232f79797b2775c9bb7891013f\nRelated-Bug: #1877408\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/08605e94a411d3dabcff63f40b71d6e34bb3ed72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/08605e94a411d3dabcff63f40b71d6e34bb3ed72"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
