)]}'
{"id":"openstack%2Fovsdbapp~728880","triplet_id":"openstack%2Fovsdbapp~master~I2431f68164be5a9bbb067cfb7a91e000cbf5be1c","project":"openstack/ovsdbapp","branch":"master","topic":"bug/1879320","hashtags":[],"change_id":"I2431f68164be5a9bbb067cfb7a91e000cbf5be1c","subject":"Change \"fixture\" module names in testing","status":"ABANDONED","created":"2020-05-18 12:57:17.000000000","updated":"2020-07-13 17:03:13.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6323d44ef2f849bc60fd8bcd46426277d703d569","_number":728880,"virtual_id_number":728880,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-17 09:21:09.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-20 12:35:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-17 08:55:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2020-07-07 19:57:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-20 12:35:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-17 08:55:15.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2020-07-07 19:57:14.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"a9efdcc0b07757323fc355a41c6ff9873a3cb6a9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-18 12:57:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20913936e54ca0c604a65e5a0b9727d900e380b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 15:01:28.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22d05819d4a3445ea5a39f2bd7aebede : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a94c0d4f1145d7804b13e771a0aeaa : FAILURE in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cf7eac12b9b4c51b15af682c382586b : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0201dc05d8474308bb42d8b078defda1 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f092c6ceb69b49ca83a55d6a5ef5b14c : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d326ac6ab42841bb9cce16958a57a6bd : SUCCESS in 7m 28s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/04533b2d08694266bd71b12ad044d559 : SUCCESS in 5m 25s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f70f06c9bc5d4f4f9639fc99a82f8704 : SUCCESS in 1h 59m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3f3490d5bad3f7232565fe277aa8b8047bee1cf2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-05-20 10:41:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e31d682589343a3d4965761d8d67de85d49e90c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 12:35:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f97e2ce5d1ca4e999e29d268a341b640 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a38cc9698a74b74a1bec0283d7970ad : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f3d872bdff34b39aa0c6d65b9623a77 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33807ed4948644768a67a19b2fb7f25e : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbf3e5eb200042f0a160d613f6d6cdbd : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7d1a52ed10243c3ac5bd0b2dcb220ee : SUCCESS in 8m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/e1133a5995ed472883045048693bacbf : SUCCESS in 7m 47s\n- ovsdbapp-neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/4ff9c137660e4878acb5a3188333abc0 : SUCCESS in 1h 52m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4451ad3dec0cb0cdda3ed9b989223a35e7843509","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-16 17:51:50.000000000","message":"Patch Set 2:\n\nping fellow reviewers","accounts_in_message":[],"_revision_number":2},{"id":"7f3ac790078eaa490b7396cac10231d6c97e7ff3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2020-06-17 08:55:15.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\nThe only concern I have about this patch is that other projects may import the fixtures with the former path and this patch is gonna break them. We don\u0027t do that in OpenStack tho - http://codesearch.openstack.org/?q\u003dovsdbapp.tests\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d","accounts_in_message":[],"_revision_number":2},{"id":"a4adeeb57e7af32dd00bfb960ac2ead7f622ca81","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-06-17 09:21:09.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n\u003e (1 comment)\n \u003e \n \u003e The only concern I have about this patch is that other projects may\n \u003e import the fixtures with the former path and this patch is gonna\n \u003e break them. We don\u0027t do that in OpenStack tho - http://codesearch.openstack.org/?q\u003dovsdbapp.tests\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nGood point. I search for \"ovsdbapp.tests.functional.schema\" because those four changed files are under this directory [1]. Only ovsdbapp is importing those files.\n\n[1]http://codesearch.openstack.org/?q\u003dovsdbapp.tests.functional.schema\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d","accounts_in_message":[],"_revision_number":2},{"id":"47e4d96e74164bb1b4c369000d885d5c6c1d9e73","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-07-07 19:57:14.000000000","message":"Patch Set 2:\n\nI don\u0027t really like the idea of revamping the layout of the project just to work around a bug in an IDE...","accounts_in_message":[],"_revision_number":2},{"id":"0ca0d3d1891a809dc26945b8edad73c17b61c554","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-07-13 17:03:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bfd9d56cac81cccac24c57516713a92efd25f84a","revisions":{"04f29abfe026f5c2d56b289643bb2cac89898904":{"kind":"REWORK","_number":1,"created":"2020-05-18 12:57:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/80/728880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/80/728880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/728880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/728880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/728880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/80/728880/1"}}},"commit":{"parents":[{"commit":"b93b1b0c7226496fb408ec7543c6d44d2540032d","subject":"Merge \"Fix hacking min version to 3.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/b93b1b0c7226496fb408ec7543c6d44d2540032d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-18 12:57:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-18 12:57:00.000000000","tz":0},"subject":"Change \"fixture\" module names in testing","message":"Change \"fixture\" module names in testing\n\nPycharm test runner do not respect the import preference when\nimporting a module. By default in Python 3 only absolute imports\nand explicit relative are supported [1]. However, when using the\nPycharm unittest runner, the local \"fixture\" module is imported\ninstead of [2].\n\nTo avoid this problem, this patch modifies the name of the \"fixture\"\nmodules in the testing directory.\n\n\"absolute_import\" directive is removed because now is the default\nbehavior.\n\n[1]https://www.python.org/dev/peps/pep-0404/#imports\n[2]https://pypi.org/project/fixtures/\n\nChange-Id: I2431f68164be5a9bbb067cfb7a91e000cbf5be1c\nCloses-Bug: #1879320\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/04f29abfe026f5c2d56b289643bb2cac89898904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/04f29abfe026f5c2d56b289643bb2cac89898904"}]},"branch":"refs/heads/master"},"bfd9d56cac81cccac24c57516713a92efd25f84a":{"kind":"REWORK","_number":2,"created":"2020-05-20 10:41:34.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/80/728880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/80/728880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/728880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/728880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/80/728880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/80/728880/2"}}},"commit":{"parents":[{"commit":"7736bac7dc093abbcb6990cf84d46c2fb6beeb9d","subject":"Merge \"Add indexed column support to ovsdbapp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/7736bac7dc093abbcb6990cf84d46c2fb6beeb9d"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-18 12:57:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-05-20 10:41:21.000000000","tz":0},"subject":"Change \"fixture\" module names in testing","message":"Change \"fixture\" module names in testing\n\nPycharm test runner do not respect the import preference when\nimporting a module. By default in Python 3 only absolute imports\nand explicit relative are supported [1]. However, when using the\nPycharm unittest runner, the local \"fixture\" module is imported\ninstead of [2].\n\nTo avoid this problem, this patch modifies the name of the \"fixture\"\nmodules in the testing directory.\n\n\"absolute_import\" directive is removed because now is the default\nbehavior.\n\n[1]https://www.python.org/dev/peps/pep-0404/#imports\n[2]https://pypi.org/project/fixtures/\n\nChange-Id: I2431f68164be5a9bbb067cfb7a91e000cbf5be1c\nCloses-Bug: #1879320\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bfd9d56cac81cccac24c57516713a92efd25f84a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/bfd9d56cac81cccac24c57516713a92efd25f84a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
