)]}'
{"id":"openstack%2Fovsdbapp~729360","triplet_id":"openstack%2Fovsdbapp~stable%2Ftrain~I9c7994d50066adf02b0066cb8efb1beb61d2e1f0","project":"openstack/ovsdbapp","branch":"stable/train","hashtags":[],"change_id":"I9c7994d50066adf02b0066cb8efb1beb61d2e1f0","subject":"Add indexed column support to ovsdbapp","status":"ABANDONED","created":"2020-05-19 19:14:33.000000000","updated":"2021-02-22 17:31:06.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"04d67d0654bc5de785001980a68a389ffd97c403","_number":729360,"virtual_id_number":729360,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"date":"2020-05-31 08:35:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-19 21:42:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":-1,"date":"2020-06-01 03:37:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-19 21:42:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-31 08:35:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-11-24 15:43:13.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"6a880b13231eb7c71178b8cbe156f174af1d3b3d","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 19:14:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"934333f2cf7b681994d5c75ed9a6cb45b8a49284","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-05-19 20:18:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6a1ac05099829b3b451bac4826a6e4fc0f9bdccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 21:42:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24481d088654ecc9e1d3ffe0c5df4df : SUCCESS in 8m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e89b4a43ff9941a48e89ee21798403f3 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9535fd3ccf94db68118af245fcbeb42 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87d8ddb5af3d48e29dbef84975277d74 : SUCCESS in 6m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ce2840085f34166864ef246110335da : SUCCESS in 7m 57s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/62dc23f598d54336afd14292aab5dd7c : SUCCESS in 9m 36s\n- ovsdbapp-tempest-dsvm-networking-ovn-ovs-release https://zuul.opendev.org/t/openstack/build/90639b165c9c40c8adf214b0fb64fd25 : SUCCESS in 1h 17m 10s","accounts_in_message":[],"_revision_number":2},{"id":"bae627e4159ff2c1e5016f712e380c46dadad924","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-31 08:35:32.000000000","message":"Patch Set 2:\n\nplease do not approve until https://bugs.launchpad.net/neutron/+bug/1881424 is resolved without need to revert","accounts_in_message":[],"_revision_number":2},{"id":"cb38e6935d98933294902b89e8972b3a1d3923d9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-06-01 03:37:09.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"85f022e508209a25de404a1791b205e7b06ece64","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2020-11-19 18:12:22.000000000","message":"Patch Set 2:\n\nmost likely not going to backport this upstream due to the tiny API change that almost certainly won\u0027t break anything after we fixed neutron, but that I can\u0027t be sure about.","accounts_in_message":[],"_revision_number":2},{"id":"a5d71baa1d64e91491e7a362ca25397c2c30a57d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-11-24 15:43:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04d67d0654bc5de785001980a68a389ffd97c403","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2021-02-22 17:31:06.000000000","message":"Abandoned\n\nToo big a change for upstream backporting. Potential for breakage.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"620aa67a114394da4c78bd4aba5a92cf0b506030","revisions":{"766414a23d78edc09bd152771a11d9107ca494ff":{"kind":"REWORK","_number":1,"created":"2020-05-19 19:14:33.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/60/729360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/60/729360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/60/729360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/60/729360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/60/729360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/60/729360/1"}}},"commit":{"parents":[{"commit":"9bb0eee28608d554076a1d9273c607afe0a5e0c3","subject":"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9bb0eee28608d554076a1d9273c607afe0a5e0c3"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 19:14:33.000000000","tz":0},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nThere are slight API changes w/ Backend.ovsdb_connection\nproperties added and start_connection becoming an instance method\ninstead of a classmethod. But since ovsdb_connection was the\nitem being stored on the class and it still is via an overridable\nproperty, users shouldn\u0027t be broken by the change.\n\nThe backport disables auto_index by default\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n(cherry picked from commit db0cdd9a4ffd64d99a6b012b5d0bf2c5bd4f96f4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/766414a23d78edc09bd152771a11d9107ca494ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/766414a23d78edc09bd152771a11d9107ca494ff"}]},"branch":"refs/heads/stable/train"},"620aa67a114394da4c78bd4aba5a92cf0b506030":{"kind":"REWORK","_number":2,"created":"2020-05-19 20:18:51.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/60/729360/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ovsdbapp","ref":"refs/changes/60/729360/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/60/729360/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/60/729360/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ovsdbapp refs/changes/60/729360/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ovsdbapp refs/changes/60/729360/2"}}},"commit":{"parents":[{"commit":"9bb0eee28608d554076a1d9273c607afe0a5e0c3","subject":"Add \"inactivity_probe\" parameter to \"enable_connection_uri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/9bb0eee28608d554076a1d9273c607afe0a5e0c3"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-07 21:34:38.000000000","tz":0},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2020-05-19 20:18:07.000000000","tz":0},"subject":"Add indexed column support to ovsdbapp","message":"Add indexed column support to ovsdbapp\n\nAdd support for python-ovs multi-column indexes. This is currently\ndisabled by default, though in a simple test of creating 10k\nLogical_Switches the time went from \"I give up after 10 mins\" to\n4 seconds. The only reason I didn\u0027t enable it by default was in\ncase users want to tailor which columns they index.\n\nThere are slight API changes w/ Backend.ovsdb_connection\nproperties added and start_connection becoming an instance method\ninstead of a classmethod. But since ovsdb_connection was the\nitem being stored on the class and it still is via an overridable\nproperty, users shouldn\u0027t be broken by the change.\n\nThe backport disables auto_index by default\n\nDepends-On: https://review.opendev.org/#/c/727232/\nChange-Id: I9c7994d50066adf02b0066cb8efb1beb61d2e1f0\n(cherry picked from commit db0cdd9a4ffd64d99a6b012b5d0bf2c5bd4f96f4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/620aa67a114394da4c78bd4aba5a92cf0b506030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ovsdbapp/commit/620aa67a114394da4c78bd4aba5a92cf0b506030"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
